)]}'
{"id":"openstack%2Fproject-config~677647","triplet_id":"openstack%2Fproject-config~master~I769f0199a2a8891f5d10f2b76bee2527dcaa3f31","project":"openstack/project-config","branch":"master","topic":"promote-starlingx","hashtags":[],"change_id":"I769f0199a2a8891f5d10f2b76bee2527dcaa3f31","subject":"Add starlingx promote jobs","status":"MERGED","created":"2019-08-21 06:41:37.000000000","updated":"2019-08-23 07:25:35.000000000","submitted":"2019-08-23 07:25:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677647-1566545135973-9d7068c6","meta_rev_id":"5bfb5d55ad5d0ec3fd65c62dfc4a0f4d9c0156ba","_number":677647,"virtual_id_number":677647,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":2,"date":"2019-08-23 07:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-08-22 16:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-08-22 15:28:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-23 07:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2019-08-22 17:06:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-23 07:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 15:28:18.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-08-22 16:59:19.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-22 17:06:06.000000000","updated_by":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"reviewer":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"state":"REVIEWER"},{"updated":"2019-08-23 07:15:17.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-08-23 07:25:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"400748c380f81df8e9c1e4f4488487f9ed61c3e1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-21 06:41:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"706d1f1b73548d52b43ac0c6d271d489daa8e6eb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-21 06:44:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0628d24aeaca390462e5ebc7a621dd5106301d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 06:58:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f802982fd18b4d829e80637753da2515 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a4690f6bbd487410ada7150384ad7279c1cb3282","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-08-22 15:28:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"58068fc04cdb3201d68ad4d5281c8ff034ec4bb1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-22 16:59:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1eb2cd9956fbc93dad865695f439a97846d1b3a9","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2019-08-22 17:06:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bc6fbf5e083545fb9a4e2b1f23ba4ec4bc85809f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-23 07:15:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f84669c6bbe8aa077963f333a2c21f263bb89cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 07:15:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f805b7ec02c3b952f9b730d97ebc9fae525e306e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 07:25:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ab104f0e5c64b70a2a1e7a4aa58abdc : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":2},{"id":"4b54b0b66aa7de4eba197dbfb2267e877cc630ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 07:25:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a648836a0035837470ed0b9a3b9fe2dc557ea58","revisions":{"8391e90fbefe59cba9282ccc420098ea96f2bb6f":{"kind":"REWORK","_number":1,"created":"2019-08-21 06:41:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/47/677647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/677647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/677647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/677647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/677647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/677647/1"}}},"commit":{"parents":[{"commit":"7c52d2e376e3c5f9e4731d52299e5355a30e18fd","subject":"Merge \"Add os_murano project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c52d2e376e3c5f9e4731d52299e5355a30e18fd"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-21 06:41:06.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-21 06:41:06.000000000","tz":120},"subject":"Add starlingx promote jobs","message":"Add starlingx promote jobs\n\nAdd new promote jobs for starling docs and api-ref.\n\nChange-Id: I769f0199a2a8891f5d10f2b76bee2527dcaa3f31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8391e90fbefe59cba9282ccc420098ea96f2bb6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8391e90fbefe59cba9282ccc420098ea96f2bb6f"}]},"branch":"refs/heads/master"},"0a648836a0035837470ed0b9a3b9fe2dc557ea58":{"kind":"REWORK","_number":2,"created":"2019-08-21 06:44:05.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/47/677647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/677647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/677647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/677647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/677647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/677647/2"}}},"commit":{"parents":[{"commit":"7c52d2e376e3c5f9e4731d52299e5355a30e18fd","subject":"Merge \"Add os_murano project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c52d2e376e3c5f9e4731d52299e5355a30e18fd"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-21 06:41:06.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-21 06:43:22.000000000","tz":120},"subject":"Add starlingx promote jobs","message":"Add starlingx promote jobs\n\nAdd new promote jobs for starling docs and api-ref.\n\nNote that starlingx does not publish to latest and neither has branch\nabbreviations setup for publish job, so follow this with promote job.\n\nChange-Id: I769f0199a2a8891f5d10f2b76bee2527dcaa3f31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a648836a0035837470ed0b9a3b9fe2dc557ea58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a648836a0035837470ed0b9a3b9fe2dc557ea58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
