)]}'
{"id":"openstack%2Fproject-config~678148","triplet_id":"openstack%2Fproject-config~master~I8328ccedbd009845ed304d6ac847cb85f02d53a4","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I8328ccedbd009845ed304d6ac847cb85f02d53a4","subject":"Rotate proposal-bot key","status":"MERGED","created":"2019-08-23 05:28:35.000000000","updated":"2019-08-23 05:49:05.000000000","submitted":"2019-08-23 05:49:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678148-1566539345930-4a7959b6","meta_rev_id":"322e5a870f35412229af00f5df5937c29cb5d201","_number":678148,"virtual_id_number":678148,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-23 05:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-23 05:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-23 05:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 05:49:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c55382b374a0164b8e84a5b26b759bd1301cfbb5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-23 05:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f169bf9cb5b9c8999da9f21291710660cb7b3563","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-23 05:29:55.000000000","message":"Patch Set 1:\n\nNote I have added this key to proposal-bot\u0027s user\n\n $ ssh -i ~/proposal-key/proposal-bot -p 29418 proposal-bot@review.opendev.org\n\n  ****    Welcome to Gerrit Code Review    ****\n\n  Hi OpenStack Proposal Bot, you have successfully connected over SSH.","accounts_in_message":[],"_revision_number":1},{"id":"d5a06d5cf24abf81b4d58601e5c95b34da1ac68c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-23 05:32:51.000000000","message":"Patch Set 1:\n\ni have also removed the old key","accounts_in_message":[],"_revision_number":1},{"id":"889fcd6ead9369c31d189276e9f66c5043209d97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:39:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/765fba3eecdc47e4ad874e3ec7cc36ba : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"27c9f4d2c8beea4b1e05e1aa6fecb78ec5669983","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-23 05:40:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nself approving as a follow-on to Iac9054de55728601d839c1feede15c2a5acc45b6","accounts_in_message":[],"_revision_number":1},{"id":"2a00a504a0b4728b5f2c286ea659916616bc14e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:41:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"888c8460cc227ab3413d2d8a972864ac646a0ca6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:49:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1092f6792e184d9c86b16068edae1664 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"322e5a870f35412229af00f5df5937c29cb5d201","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:49:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15bd17a6a3751a868a6258279ca551fca8fa9c8d","revisions":{"15bd17a6a3751a868a6258279ca551fca8fa9c8d":{"kind":"REWORK","_number":1,"created":"2019-08-23 05:28:35.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/48/678148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/678148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/678148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/678148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/678148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/678148/1"}}},"commit":{"parents":[{"commit":"e1d227bf5b70f531ca608202d47d5797536c67ad","subject":"Move external propose-updates changes to project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1d227bf5b70f531ca608202d47d5797536c67ad"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-23 05:27:54.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-23 05:27:54.000000000","tz":600},"subject":"Rotate proposal-bot key","message":"Rotate proposal-bot key\n\nChange-Id: I8328ccedbd009845ed304d6ac847cb85f02d53a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15bd17a6a3751a868a6258279ca551fca8fa9c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15bd17a6a3751a868a6258279ca551fca8fa9c8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
