)]}'
{"id":"openstack%2Fproject-config~678214","triplet_id":"openstack%2Fproject-config~master~I0e5792741ee6b3a54a957c30fd6f8b8a0d2471cd","project":"openstack/project-config","branch":"master","topic":"update-tc-voting-rules","hashtags":[],"change_id":"I0e5792741ee6b3a54a957c30fd6f8b8a0d2471cd","subject":"Allow all TC members +W access","status":"MERGED","created":"2019-08-23 12:35:02.000000000","updated":"2019-09-11 16:17:05.000000000","submitted":"2019-09-11 16:17:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678214-1568218625695-1839ade1","meta_rev_id":"b20e9d03ddbc2a1711133d9e22e74ccad9201c18","_number":678214,"virtual_id_number":678214,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-11 16:17:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-09-11 16:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-11 14:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-09-11 16:03:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 12:35:54.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-09-11 14:49:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-11 16:03:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-11 16:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2730e6a6209805fa7f17d878809674c8bce324da","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-23 12:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18bb109d0cf1cad793d7706aa4367c394bf5fd58","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-08-23 12:35:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d234726379000c5357f7a8d2212780594c4059a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 12:42:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f2f9ceb3b7d54e6fa0d666f0fdeedee6 : SUCCESS in 7m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e6d10a802a5047579f771cbc8da740e8 : FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"acbfa84fe6d4c6491c8153ffbf4aae7cbcea11f5","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-23 13:33:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41f2914c3d9c6998bd52eedc2b27bdd175460bf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:46:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/361527512d0f41c6a7630d75ff953df5 : SUCCESS in 7m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/91c8d7087c6b43a69c1fd6be084fd0ad : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4d1c4ab6dcd8666ea3c950cbaf497c3d66770073","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-09-11 14:39:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f830d4e3a2997f8dfb0d8938c9665c2dd352d208","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-11 14:49:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ff4872fa151818e6ae5c53082444b7aadb7ac52a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 14:52:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0be76dd7d1f64f0599e2a3a0b3678b81 : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f3e60d5c4d1c450b8f5a12e9f2042d87 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"3f1c886ee49814b0fa23c3cb0d5bc10ccce72d24","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-11 16:03:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"32c76bc3dd750404d6642c88a9fd0f8d6a2f653c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:03:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ae535a6141f72f56493e44dc31fc980cb6118996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:17:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2d16d71bf5c213a8fe706bfac03171d809f1bc96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:17:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26f1f6d10d7b44a5a8bc6b323c01557e : SUCCESS in 7m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dee4d425b81e4be8ba65efbac35f31c7 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d9e2fc6c065ea62cb89a98c42d191f7df358567","revisions":{"8ee794faa8aed9839ac8432b740f85b372948130":{"kind":"REWORK","_number":1,"created":"2019-08-23 12:35:02.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/14/678214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/678214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/678214/1"}}},"commit":{"parents":[{"commit":"8f84c98a0c0637478ea93af88b24ceba36d71d68","subject":"Revert SSH fingerprint of review.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f84c98a0c0637478ea93af88b24ceba36d71d68"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 12:34:26.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 12:35:00.000000000","tz":60},"subject":"Allow all TC members +W access","message":"Allow all TC members +W access\n\nDepends-on: https://review.opendev.org/678212\n\nChange-Id: I0e5792741ee6b3a54a957c30fd6f8b8a0d2471cd\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ee794faa8aed9839ac8432b740f85b372948130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ee794faa8aed9839ac8432b740f85b372948130"}]},"branch":"refs/heads/master"},"a6ec8e4cb2d7239723070d6f6d3a1c77ce794cbd":{"kind":"REWORK","_number":2,"created":"2019-08-23 13:33:26.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/14/678214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/678214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/678214/2"}}},"commit":{"parents":[{"commit":"8f84c98a0c0637478ea93af88b24ceba36d71d68","subject":"Revert SSH fingerprint of review.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f84c98a0c0637478ea93af88b24ceba36d71d68"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 12:34:26.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 13:33:18.000000000","tz":60},"subject":"Allow all TC members +W access","message":"Allow all TC members +W access\n\nDepends-on: https://review.opendev.org/678212\n\nChange-Id: I0e5792741ee6b3a54a957c30fd6f8b8a0d2471cd\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6ec8e4cb2d7239723070d6f6d3a1c77ce794cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6ec8e4cb2d7239723070d6f6d3a1c77ce794cbd"}]},"branch":"refs/heads/master"},"6d9e2fc6c065ea62cb89a98c42d191f7df358567":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-11 14:39:48.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/14/678214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/678214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/678214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/678214/3"}}},"commit":{"parents":[{"commit":"25f2e102aa6288d6d1536fbccad2299e0068b1a9","subject":"Merge \"Add Fedora 30 nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25f2e102aa6288d6d1536fbccad2299e0068b1a9"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 12:34:26.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-09-11 14:39:34.000000000","tz":60},"subject":"Allow all TC members +W access","message":"Allow all TC members +W access\n\nDepends-on: https://review.opendev.org/678212\n\nChange-Id: I0e5792741ee6b3a54a957c30fd6f8b8a0d2471cd\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d9e2fc6c065ea62cb89a98c42d191f7df358567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d9e2fc6c065ea62cb89a98c42d191f7df358567"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
