)]}'
{"id":"openstack%2Fproject-config~678344","triplet_id":"openstack%2Fproject-config~master~Ia970c18f5fcbdf69e02412b7cf957c9387daad19","project":"openstack/project-config","branch":"master","topic":"promote-docs","hashtags":[],"change_id":"Ia970c18f5fcbdf69e02412b7cf957c9387daad19","subject":"Add PDF to api promote job","status":"MERGED","created":"2019-08-24 10:39:59.000000000","updated":"2019-08-26 16:53:56.000000000","submitted":"2019-08-26 16:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678344-1566838436615-6ec52b8c","meta_rev_id":"1b17196f9eea6d69fb4852f59ce885eabbd626ee","_number":678344,"virtual_id_number":678344,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-26 16:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-08-26 16:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-26 06:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-08-26 16:37:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 06:18:10.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-08-26 16:37:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-26 16:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1b28724c99cb385fa22e089a67eef3bfd1c7f54","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 10:39:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cbbafd09caeffd27bae42c22635288e1521ccc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 10:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e5d15220d20648deab482a691882af42 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d9b88ac887ec32d9ccf65cea72f3ed6eb27c731d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-26 05:47:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0224509d00acf1e0fbe0b412fcf24a2803edc95","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 06:16:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78fcae1c775b469d865cfb505718f291d3840c25","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-26 06:18:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c005971450de26bb722bcac342120a00496e29ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 06:29:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f9c9fe0341024aa3918c1a9ba214f2b0 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9c54ea2ad71cf97ad4b99ada359a5a27125ba436","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-26 16:37:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c48cde0f3cf428c60ab051bbc2d1a1d4a846127f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:38:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"444a6608b9c7b2a962c5939a7e879feee93c40da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:53:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be84f30833ea41d9958cacd2f92f48b8 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":2},{"id":"88f5f8fb41b6ec8dbca9ea8e960624708f5d25ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:53:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d12c24cbcf062fd29939a4b93cb6afb1f1f2bb7d","revisions":{"00594be4f562cfa5a9daf7aeef425cf5db7b8daf":{"kind":"REWORK","_number":1,"created":"2019-08-24 10:39:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/44/678344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/678344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/678344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/678344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/678344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/678344/1"}}},"commit":{"parents":[{"commit":"90615f9aa72e2cc8a1e450bfde3f678e96135a65","subject":"Remove now obsolete publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90615f9aa72e2cc8a1e450bfde3f678e96135a65"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 10:39:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 10:39:04.000000000","tz":120},"subject":"Add PDF to api promote job","message":"Add PDF to api promote job\n\nCopy over Ic0043bebf902a24957ead1b6d1f9c05d5c834768 for this specific\npromote file.\n\nChange-Id: Ia970c18f5fcbdf69e02412b7cf957c9387daad19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00594be4f562cfa5a9daf7aeef425cf5db7b8daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00594be4f562cfa5a9daf7aeef425cf5db7b8daf"}]},"branch":"refs/heads/master"},"d12c24cbcf062fd29939a4b93cb6afb1f1f2bb7d":{"kind":"REWORK","_number":2,"created":"2019-08-26 06:16:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/44/678344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/678344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/678344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/678344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/678344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/678344/2"}}},"commit":{"parents":[{"commit":"90615f9aa72e2cc8a1e450bfde3f678e96135a65","subject":"Remove now obsolete publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90615f9aa72e2cc8a1e450bfde3f678e96135a65"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 10:39:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-26 06:12:34.000000000","tz":120},"subject":"Add PDF to api promote job","message":"Add PDF to api promote job\n\nCopy over Ic0043bebf902a24957ead1b6d1f9c05d5c834768 for this specific\npromote file.\n\nChange-Id: Ia970c18f5fcbdf69e02412b7cf957c9387daad19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d12c24cbcf062fd29939a4b93cb6afb1f1f2bb7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d12c24cbcf062fd29939a4b93cb6afb1f1f2bb7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
