)]}'
{"id":"openstack%2Fproject-config~678356","triplet_id":"openstack%2Fproject-config~master~I5cc73dc0c3a25e803a555c045ad32c3197e798a9","project":"openstack/project-config","branch":"master","topic":"promote-docs","hashtags":[],"change_id":"I5cc73dc0c3a25e803a555c045ad32c3197e798a9","subject":"Remove now unused publish jobs","status":"MERGED","created":"2019-08-24 14:26:59.000000000","updated":"2019-09-10 06:42:40.000000000","submitted":"2019-09-10 06:42:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678356-1568097760652-3ba29265","meta_rev_id":"a389fe8a840c4444722ceef79bedea0a51f98235","_number":678356,"virtual_id_number":678356,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-09 08:08:52.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-09-10 06:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-09 14:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-10 06:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-10 06:34:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-09 14:14:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-09-10 06:34:08.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-09-10 06:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"211e6b454b12ad012afb86608f3f58920a035721","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 14:26:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eda37c217bde96d019c04dc41d519190bf54e355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 14:27:31.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"4a88049a9c31b0f1520b05ea752ee5a6e7043fd0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 15:08:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f115fc3cea2e07f8e7998e8749aeadb7ec34767d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 15:08:48.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"763c584690b5c8743639f2efa21bef6b6360d07c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 15:17:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a9257e3d72cc0d3cecf6345ac24910c28cf72930","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 15:18:40.000000000","message":"Patch Set 3:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"8972f9957d15163ec34d3260da81365199301e31","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 06:17:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"29d5cb9dd1975f1f9bf5cf6b3563ba70c2a5f6fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 06:17:58.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"d76b591a03cc07f23a29e035228fa5efd4bdab89","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 16:31:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"82138a37510c55877208450504a8ccf13ffc9787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:32:39.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":5},{"id":"d333b82b3446ccc1dd5cddb9a7ac07a688597512","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 18:09:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0c1fb516235a13d05b2b772715a7488b2208a37a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:11:03.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ossa on branch master.  The error was:\n\n  Job publish-security not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - build-openstack-docs-pti\n      post:\n        jobs:\n          - publish-security\n\n  in \"openstack/ossa/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":5},{"id":"00cf625eeafe2de0424cc23fac78fe647abd40f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-09 08:08:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e9b7064277db1162066b327861d8b1d920e5d2ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 08:30:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b1d50db98e1446a99e3ea7cc6b8dc15 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":5},{"id":"18b8b75b1133f728aabb9de0de9d1f9ac93feac2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-09 14:14:29.000000000","message":"Patch Set 5: Code-Review+2\n\nStill has one unmerged dep.","accounts_in_message":[],"_revision_number":5},{"id":"39367b0568bd5735c861e536d8e87721f3a513fd","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-10 06:34:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5b129715b3c39f63624b406a513f6a41d73bf49f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 06:34:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7e6a6353ab96e7e8ce1c3ee29516c0dd26195d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 06:42:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d053808221745f8bd58ab119490b560 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":5},{"id":"30b223d2b6ccc0cea5398150a309a4521529f5b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 06:42:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"88a4c7d78538fab06b68ee2dcaec407a762db5f9","revisions":{"bd147efbc6d4cd5a4893d2085d9b631d680cabcb":{"kind":"REWORK","_number":1,"created":"2019-08-24 14:26:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/56/678356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/678356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/678356/1"}}},"commit":{"parents":[{"commit":"8dfaece81ce1a93f9240d424ffebc84eb8a1298a","subject":"Add promote jobs for static site","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8dfaece81ce1a93f9240d424ffebc84eb8a1298a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 14:20:05.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 14:26:39.000000000","tz":120},"subject":"Remove now unused publish jobs","message":"Remove now unused publish jobs\n\nA couple of repos now use promote jobs instead of publish jobs, remove\nthe now unused promote jobs.\n\nDepends-On: https://review.opendev.org/678352\nDepends-On: https://review.opendev.org/678353\nDepends-On: https://review.opendev.org/678355\nChange-Id: I5cc73dc0c3a25e803a555c045ad32c3197e798a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd147efbc6d4cd5a4893d2085d9b631d680cabcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd147efbc6d4cd5a4893d2085d9b631d680cabcb"}]},"branch":"refs/heads/master"},"645293d961104b215c842c92f4bdfbeefe7e8a94":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-25 15:08:06.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/56/678356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/678356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/678356/2"}}},"commit":{"parents":[{"commit":"03f4ce7276a5fe8ab575d842236361ed13fa139b","subject":"Add promote jobs for static site / releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03f4ce7276a5fe8ab575d842236361ed13fa139b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 14:20:05.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-25 15:07:08.000000000","tz":120},"subject":"Remove now unused publish jobs","message":"Remove now unused publish jobs\n\nA couple of repos now use promote jobs instead of publish jobs, remove\nthe now unused promote jobs.\n\nDepends-On: https://review.opendev.org/678352\nDepends-On: https://review.opendev.org/678353\nDepends-On: https://review.opendev.org/678355\nChange-Id: I5cc73dc0c3a25e803a555c045ad32c3197e798a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/645293d961104b215c842c92f4bdfbeefe7e8a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/645293d961104b215c842c92f4bdfbeefe7e8a94"}]},"branch":"refs/heads/master"},"e5fcc759441557e9a22d2a3bc77577498f20ba9b":{"kind":"NO_CHANGE","_number":3,"created":"2019-08-25 15:17:49.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/56/678356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/678356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/678356/3"}}},"commit":{"parents":[{"commit":"a1f185234af679edcf456f8e4553d352b1eaf045","subject":"Add promote jobs for static site / releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1f185234af679edcf456f8e4553d352b1eaf045"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 14:20:05.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-25 15:17:37.000000000","tz":120},"subject":"Remove now unused publish jobs","message":"Remove now unused publish jobs\n\nA couple of repos now use promote jobs instead of publish jobs, remove\nthe now unused promote jobs.\n\nDepends-On: https://review.opendev.org/678352\nDepends-On: https://review.opendev.org/678353\nDepends-On: https://review.opendev.org/678355\nChange-Id: I5cc73dc0c3a25e803a555c045ad32c3197e798a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e5fcc759441557e9a22d2a3bc77577498f20ba9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e5fcc759441557e9a22d2a3bc77577498f20ba9b"}]},"branch":"refs/heads/master"},"d173e7c4928b9df65f461850a974587e975672b0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-26 06:17:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/56/678356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/678356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/678356/4"}}},"commit":{"parents":[{"commit":"c7033acae97fbfcc7cf9e111eb00b09cb4716782","subject":"Add promote jobs for static site / releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7033acae97fbfcc7cf9e111eb00b09cb4716782"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 14:20:05.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-26 06:17:24.000000000","tz":0},"subject":"Remove now unused publish jobs","message":"Remove now unused publish jobs\n\nA couple of repos now use promote jobs instead of publish jobs, remove\nthe now unused promote jobs.\n\nDepends-On: https://review.opendev.org/678352\nDepends-On: https://review.opendev.org/678353\nDepends-On: https://review.opendev.org/678355\nChange-Id: I5cc73dc0c3a25e803a555c045ad32c3197e798a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d173e7c4928b9df65f461850a974587e975672b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d173e7c4928b9df65f461850a974587e975672b0"}]},"branch":"refs/heads/master"},"88a4c7d78538fab06b68ee2dcaec407a762db5f9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-26 16:31:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/56/678356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/678356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/678356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/678356/5"}}},"commit":{"parents":[{"commit":"ef1834e2ab36bd86bb3a93ab6fcc2093fdcdb167","subject":"Add promote jobs for static site / releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef1834e2ab36bd86bb3a93ab6fcc2093fdcdb167"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 14:20:05.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-26 16:31:28.000000000","tz":120},"subject":"Remove now unused publish jobs","message":"Remove now unused publish jobs\n\nA couple of repos now use promote jobs instead of publish jobs, remove\nthe now unused promote jobs.\n\nDepends-On: https://review.opendev.org/678352\nDepends-On: https://review.opendev.org/678353\nDepends-On: https://review.opendev.org/678355\nChange-Id: I5cc73dc0c3a25e803a555c045ad32c3197e798a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a4c7d78538fab06b68ee2dcaec407a762db5f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a4c7d78538fab06b68ee2dcaec407a762db5f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
