)]}'
{"id":"openstack%2Fproject-config~678385","triplet_id":"openstack%2Fproject-config~master~Ic7352e49dd8a1ec862d996f32de918f90a8084de","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ic7352e49dd8a1ec862d996f32de918f90a8084de","subject":"Retire swauth - remove project","status":"MERGED","created":"2019-08-24 18:05:24.000000000","updated":"2019-08-28 13:54:06.000000000","submitted":"2019-08-28 13:54:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678385-1567000446259-2a78a105","meta_rev_id":"0e07d872c5d0a9796cdb8a2c177445510d14e3c6","_number":678385,"virtual_id_number":678385,"owner":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-28 13:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},{"value":2,"date":"2019-08-27 22:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-24 18:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-08-28 13:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 22:04:18.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-28 13:40:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-28 13:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afe0501c14c412d31013fb4c94437693a5650658","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 18:05:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54967c6238b48d4dbdf17ab3a3aca70cd13baff8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 18:07:18.000000000","message":"Patch Set 1: Code-Review-1\n\nremove it from zuul/main.yaml as well","accounts_in_message":[],"_revision_number":1},{"id":"b2c4bafe6133d0371b86509838425d4de9ad86ea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 18:07:45.000000000","message":"Patch Set 1:\n\nand update gerrit/projects.yaml, this is all explained in the Infra Manual...","accounts_in_message":[],"_revision_number":1},{"id":"395d84c2618d5cb711c8d840301a56ec38eba535","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 18:09:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"918df5447bdc5bd0ca51431085ec968d4c9d4204","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 18:10:26.000000000","message":"Patch Set 2:\n\nbefore we can merge this, please abandon all open requests (besides the retirement one), see https://review.opendev.org/#/q/project:x/swauth+is:open","accounts_in_message":[],"_revision_number":2},{"id":"ca9869a7285e0425cd20b4b0b738a1a9876f28b2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 18:11:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40f1f9db948b7b5d5c8d929d6ba637f12d177153","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 18:13:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ded14313daad7aaac943efdcc505dd59f1c13a8","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 18:14:16.000000000","message":"Patch Set 2:\n\n\u003e before we can merge this, please abandon all open requests (besides\n \u003e the retirement one), see https://review.opendev.org/#/q/project:x/swauth+is:open\n\ndone","accounts_in_message":[],"_revision_number":2},{"id":"e2a2fd90bfeddf03ba553d224a00caa533c7c407","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 18:14:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb6d6224e562c66276330df71985be814490a7fc","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 18:21:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1eeadee6648ddd02e3848fb29b422c6d19fa24c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 18:33:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2f6c05887f754438a4651cd9e1e94f04 : SUCCESS in 6m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d64c8fed7cb349aa898d8ace2a78f3eb : SUCCESS in 4m 29s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/cc59c9159b534ce0a9032fc0e28ffc0e : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":4},{"id":"0c5c88706768f794866f2172821feba6aa8b96bc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 18:44:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0ab13c90fa548f533717c0187e025239b26e9e12","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-27 22:04:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9d419c98ced16b22d7d470bf696d0cac7cd0ac42","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-28 13:40:29.000000000","message":"Patch Set 4: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":4},{"id":"b3abc98281910fd8061e5c228b586d704d511170","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:40:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3613ebab9fc682107e7dcb144adb7706b9bca1db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:54:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/61df5e077387420f93fc54c149521e95 : SUCCESS in 8m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/36f7f532684549b6a43f8dfe58f22357 : SUCCESS in 4m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c0c930c603f04a378ab2beca0aa194ba : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":4},{"id":"0e07d872c5d0a9796cdb8a2c177445510d14e3c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:54:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0d526ddaa9197724911d0b267eb21e00f212f5f6","revisions":{"5254e7ab84bb295b477b87b8ce6f75d43d2e1328":{"kind":"REWORK","_number":1,"created":"2019-08-24 18:05:24.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/85/678385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/678385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/678385/1"}}},"commit":{"parents":[{"commit":"2600a4c5b356e4777b68f2717cd34c3ab5e35db0","subject":"Retire swauth - stop gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2600a4c5b356e4777b68f2717cd34c3ab5e35db0"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:03:26.000000000","tz":120},"committer":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:03:26.000000000","tz":120},"subject":"Retire swauth - remove project","message":"Retire swauth - remove project\n\nDepends-On: https://review.opendev.org/678378\nChange-Id: Ic7352e49dd8a1ec862d996f32de918f90a8084de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5254e7ab84bb295b477b87b8ce6f75d43d2e1328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5254e7ab84bb295b477b87b8ce6f75d43d2e1328"}]},"branch":"refs/heads/master"},"f6c923f0ba3c05694ca40ff5e69f838a1138f43a":{"kind":"REWORK","_number":2,"created":"2019-08-24 18:09:40.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/85/678385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/678385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/678385/2"}}},"commit":{"parents":[{"commit":"2600a4c5b356e4777b68f2717cd34c3ab5e35db0","subject":"Retire swauth - stop gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2600a4c5b356e4777b68f2717cd34c3ab5e35db0"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:03:26.000000000","tz":120},"committer":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:09:11.000000000","tz":120},"subject":"Retire swauth - remove project","message":"Retire swauth - remove project\n\nDepends-On: https://review.opendev.org/678378\nChange-Id: Ic7352e49dd8a1ec862d996f32de918f90a8084de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6c923f0ba3c05694ca40ff5e69f838a1138f43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6c923f0ba3c05694ca40ff5e69f838a1138f43a"}]},"branch":"refs/heads/master"},"96f547bedadbe5c748e877aeb1029946a47d3889":{"kind":"REWORK","_number":3,"created":"2019-08-24 18:14:34.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/85/678385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/678385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/678385/3"}}},"commit":{"parents":[{"commit":"2600a4c5b356e4777b68f2717cd34c3ab5e35db0","subject":"Retire swauth - stop gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2600a4c5b356e4777b68f2717cd34c3ab5e35db0"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:03:26.000000000","tz":120},"committer":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:13:34.000000000","tz":120},"subject":"Retire swauth - remove project","message":"Retire swauth - remove project\n\nDepends-On: https://review.opendev.org/678378\nChange-Id: Ic7352e49dd8a1ec862d996f32de918f90a8084de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96f547bedadbe5c748e877aeb1029946a47d3889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96f547bedadbe5c748e877aeb1029946a47d3889"}]},"branch":"refs/heads/master"},"0d526ddaa9197724911d0b267eb21e00f212f5f6":{"kind":"REWORK","_number":4,"created":"2019-08-24 18:21:42.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/85/678385/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/678385/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/678385/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/678385/4"}}},"commit":{"parents":[{"commit":"2600a4c5b356e4777b68f2717cd34c3ab5e35db0","subject":"Retire swauth - stop gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2600a4c5b356e4777b68f2717cd34c3ab5e35db0"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:03:26.000000000","tz":120},"committer":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 18:21:03.000000000","tz":120},"subject":"Retire swauth - remove project","message":"Retire swauth - remove project\n\nDepends-On: https://review.opendev.org/678378\nChange-Id: Ic7352e49dd8a1ec862d996f32de918f90a8084de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0d526ddaa9197724911d0b267eb21e00f212f5f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0d526ddaa9197724911d0b267eb21e00f212f5f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
