)]}'
{"id":"openstack%2Fproject-config~678722","triplet_id":"openstack%2Fproject-config~master~Ib45980849177494fdc24cca9a7f874abcfa827cb","project":"openstack/project-config","branch":"master","topic":"fix-static-promote","hashtags":[],"change_id":"Ib45980849177494fdc24cca9a7f874abcfa827cb","subject":"Don\u0027t use static promote jobs yet","status":"ABANDONED","created":"2019-08-27 04:48:08.000000000","updated":"2019-08-27 06:28:55.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de683c98547217aaec92a02b9b0b7cbe77413b5c","_number":678722,"virtual_id_number":678722,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-27 05:03:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 05:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc69653294b9f03b0dcbb7a4ba0d3a4989ac7416","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 04:48:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e84f4a494778323500653f1508dd34414107e48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 04:48:26.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job publish-governance-uc not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/governance-uc\n      post:\n        jobs:\n          - publish-governance-uc\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 2292, column 3","accounts_in_message":[],"_revision_number":1},{"id":"253decc56ebbdde5c7afd52abc05e27337408fc2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 04:54:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c53c745801b8e38fd0bcd6a3ea2fad62d587dfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 05:03:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee8fe543b96b4b40bfdb1879b9b3e945 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e0e43351cdc19ab0abc3c3140f0c4fc9971242fd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 06:28:55.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bed3e95896a5845053edf5751075a888355c5b2","revisions":{"aa098e13d780541b3aafddaca071411c3ea47533":{"kind":"REWORK","_number":1,"created":"2019-08-27 04:48:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/678722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/678722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/678722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/678722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/678722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/678722/1"}}},"commit":{"parents":[{"commit":"6ddcf2896032d2026dd6f5d1feb01a8843bb60f9","subject":"Fix paths for static docs playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ddcf2896032d2026dd6f5d1feb01a8843bb60f9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-27 04:47:08.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-27 04:47:08.000000000","tz":120},"subject":"Don\u0027t use static promote jobs yet","message":"Don\u0027t use static promote jobs yet\n\nRevert part of I112745b70448cc3e6fec6e2932e4fe651f9174b0: The job\nneeds further work, so don\u0027t use it yet.\n\nNote also that these are all wrongly in post queue instead of promote.\n\nChange-Id: Ib45980849177494fdc24cca9a7f874abcfa827cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa098e13d780541b3aafddaca071411c3ea47533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa098e13d780541b3aafddaca071411c3ea47533"}]},"branch":"refs/heads/master"},"7bed3e95896a5845053edf5751075a888355c5b2":{"kind":"REWORK","_number":2,"created":"2019-08-27 04:54:51.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/678722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/678722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/678722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/678722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/678722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/678722/2"}}},"commit":{"parents":[{"commit":"6ddcf2896032d2026dd6f5d1feb01a8843bb60f9","subject":"Fix paths for static docs playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ddcf2896032d2026dd6f5d1feb01a8843bb60f9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-27 04:47:08.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-27 04:54:44.000000000","tz":120},"subject":"Don\u0027t use static promote jobs yet","message":"Don\u0027t use static promote jobs yet\n\nRevert part of I112745b70448cc3e6fec6e2932e4fe651f9174b0: The job\nneeds further work, so don\u0027t use it yet.\n\nNote also that these are all wrongly in post queue instead of promote.\n\nChange-Id: Ib45980849177494fdc24cca9a7f874abcfa827cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bed3e95896a5845053edf5751075a888355c5b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bed3e95896a5845053edf5751075a888355c5b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
