)]}'
{"id":"openstack%2Fproject-config~678840","triplet_id":"openstack%2Fproject-config~master~I0c1abcd212c016bf15a18cb6f5c2554256fdb502","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I0c1abcd212c016bf15a18cb6f5c2554256fdb502","subject":"zuul: add openstack/magnum to opendev","status":"MERGED","created":"2019-08-27 13:39:35.000000000","updated":"2019-08-28 14:05:55.000000000","submitted":"2019-08-28 14:05:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678840-1567001155060-80306961","meta_rev_id":"856a5ef9b09a8b5a5a01f3bfca40f032d7414d58","_number":678840,"virtual_id_number":678840,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-08-28 14:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-08-28 13:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-08-28 13:46:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 16:01:08.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-28 09:54:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-08-28 13:46:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-28 14:05:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99e370cf4419cd3ece43f8fff90e4b1b142d8600","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:39:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b2f81242021ae85e2d28ee924fcf884f905c53d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:52:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4d857f52640d43c494c3555890371153 : SUCCESS in 7m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eda5e6f441ee4aebb89b9362a1ff09aa : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a15ea24499d1d66d9a07053f56db5ac48c8e2bba","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-27 14:17:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b16d16046322b59994be18e5628f0fc2925b47db","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-27 14:56:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db6380ddcc9a4bd75626f5d108c9841eb35289c3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-27 16:01:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"702371dd213941166196820b1cf463d1e5899c87","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 19:30:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4b7f387df988a159bb40e6bca1470e81b44936a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 19:43:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6112f0f763cb45b2a9b665bf1d2c59ed : SUCCESS in 7m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/451914a3bb2c4b8d8f8c30fa2d0874a9 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8a7b6c0627459f1bdde3abcf61eb4b527fcbb80d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-28 09:54:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7a164635bd332fdb7bac447451e0d474fe6091b7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-28 13:44:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce5656958ede57bc094f6371364c0467034fe464","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-28 13:46:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f9b88861912f894b0b8ee8534feb40f03e20b6af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-28 13:46:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nApproving after adding comment - taking frickler\u0027s +2 into account","accounts_in_message":[],"_revision_number":3},{"id":"16cc08208cf3283cc688de3125c00648c69e9025","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:56:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/245b5b86bb9e426295125a46e90a88ca : SUCCESS in 6m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b1793bdea8a84e5ab0f7619317660270 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"0d133fe0981ce7cd5bcaf4d1fcd1cc1af8d9f533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:56:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"681b7bc7d78d9aad2fa6f3b55ffaddc1d9ef07f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 14:05:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6df6cc0ae68542b7b5f8dbb69caa6833 : SUCCESS in 6m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5cd729cf16ef4bc9b07a9ebe8dcc60f5 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"856a5ef9b09a8b5a5a01f3bfca40f032d7414d58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 14:05:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"39367460e157719c2779d8212e37c4b4a0cd50f5","revisions":{"c0f927a6cf45a5014870885b1812959fde82f6f5":{"kind":"REWORK","_number":1,"created":"2019-08-27 13:39:35.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/40/678840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/678840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/678840/1"}}},"commit":{"parents":[{"commit":"b5c8fde6489e55327624652d2d1c12f686cce1f2","subject":"Fix secret for promote-stx-tox-docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5c8fde6489e55327624652d2d1c12f686cce1f2"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-27 13:38:36.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-27 13:38:36.000000000","tz":-240},"subject":"zuul: add openstack/magnum to opendev","message":"zuul: add openstack/magnum to opendev\n\nvexxhost/kue is a library that would like to make changes against\nchecked-out versions of openstack/magnum in gate (enabling us\nto use things like Depends-On).\n\nThis change adds nothing but the project in order to allow us to\nbe able to do Depends-On.\n\nChange-Id: I0c1abcd212c016bf15a18cb6f5c2554256fdb502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0f927a6cf45a5014870885b1812959fde82f6f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0f927a6cf45a5014870885b1812959fde82f6f5"}]},"branch":"refs/heads/master"},"248731cb76ffb1f70dce0d8feffced22ec7ef11f":{"kind":"REWORK","_number":2,"created":"2019-08-27 19:30:17.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/40/678840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/678840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/678840/2"}}},"commit":{"parents":[{"commit":"b5c8fde6489e55327624652d2d1c12f686cce1f2","subject":"Fix secret for promote-stx-tox-docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5c8fde6489e55327624652d2d1c12f686cce1f2"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-27 13:38:36.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-27 19:30:04.000000000","tz":-240},"subject":"zuul: add openstack/magnum to opendev","message":"zuul: add openstack/magnum to opendev\n\nvexxhost/kue is a library that would like to make changes against\nchecked-out versions of openstack/magnum in gate (enabling us\nto use things like Depends-On).\n\nThis change adds nothing but the project in order to allow us to\nbe able to do Depends-On.\n\nChange-Id: I0c1abcd212c016bf15a18cb6f5c2554256fdb502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/248731cb76ffb1f70dce0d8feffced22ec7ef11f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/248731cb76ffb1f70dce0d8feffced22ec7ef11f"}]},"branch":"refs/heads/master"},"39367460e157719c2779d8212e37c4b4a0cd50f5":{"kind":"REWORK","_number":3,"created":"2019-08-28 13:46:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/40/678840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/678840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/678840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/678840/3"}}},"commit":{"parents":[{"commit":"b5c8fde6489e55327624652d2d1c12f686cce1f2","subject":"Fix secret for promote-stx-tox-docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5c8fde6489e55327624652d2d1c12f686cce1f2"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-27 13:38:36.000000000","tz":-240},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-28 13:45:57.000000000","tz":0},"subject":"zuul: add openstack/magnum to opendev","message":"zuul: add openstack/magnum to opendev\n\nvexxhost/kue is a library that would like to make changes against\nchecked-out versions of openstack/magnum in gate (enabling us\nto use things like Depends-On).\n\nThis change adds nothing but the project in order to allow us to\nbe able to do Depends-On.\n\nChange-Id: I0c1abcd212c016bf15a18cb6f5c2554256fdb502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39367460e157719c2779d8212e37c4b4a0cd50f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39367460e157719c2779d8212e37c4b4a0cd50f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
