)]}'
{"id":"openstack%2Fproject-config~679155","triplet_id":"openstack%2Fproject-config~master~Idf25ce004619a0e6d30bce0522236c0b4d2537f8","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Idf25ce004619a0e6d30bce0522236c0b4d2537f8","subject":"Manages build reqs in alexandria project","status":"MERGED","created":"2019-08-28 22:46:27.000000000","updated":"2019-08-29 21:03:12.000000000","submitted":"2019-08-29 21:03:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679155-1567112592022-22bc38ef","meta_rev_id":"8d389eb30d89e31197522fa7fb424d73b6ac6358","_number":679155,"virtual_id_number":679155,"owner":{"_account_id":15222,"name":"Bruno Cornec","email":"bruno.cornec@hpe.com","username":"bruno-cornec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-29 21:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-08-29 20:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-29 04:53:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-08-29 20:50:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 04:53:16.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-29 20:50:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-29 21:03:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3366fabf14e01aadf8675277182ae07ae7afbc8b","author":{"_account_id":15222,"name":"Bruno Cornec","email":"bruno.cornec@hpe.com","username":"bruno-cornec"},"date":"2019-08-28 22:46:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79dc4fcc79a084f519457d65f16bae7aeca80328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 22:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/791b9b07e51f401b8e97b18511e0ae93 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dbc3adcc9c68294dfb478e65a508db9b15c46bb7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-29 04:53:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b0abdf5ceae9813c2367fecef77642ff22530aa","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-29 20:50:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e05761fce8870e6cb95d0591e1cb0d3481b17b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 20:51:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a725a02fa725be48fbc65607d9b88e1f097718dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 21:03:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dccadebf5cc443f1bd23c45521e46ee3 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8d389eb30d89e31197522fa7fb424d73b6ac6358","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 21:03:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0540284625bd678fa244d24f86ad151a42c58bc5","revisions":{"0540284625bd678fa244d24f86ad151a42c58bc5":{"kind":"REWORK","_number":1,"created":"2019-08-28 22:46:27.000000000","uploader":{"_account_id":15222,"name":"Bruno Cornec","email":"bruno.cornec@hpe.com","username":"bruno-cornec"},"ref":"refs/changes/55/679155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/679155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/679155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/679155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/679155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/679155/1"}}},"commit":{"parents":[{"commit":"0bc7cd3af9eabc326bb996d8457e4a3ab858c875","subject":"Merge \"Manages build reqs in python-redfish project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0bc7cd3af9eabc326bb996d8457e4a3ab858c875"}]}],"author":{"name":"Bruno Cornec","email":"bruno.cornec@hpe.com","date":"2019-08-28 22:44:44.000000000","tz":120},"committer":{"name":"Bruno Cornec","email":"bruno.cornec@hpe.com","date":"2019-08-28 22:44:44.000000000","tz":120},"subject":"Manages build reqs in alexandria project","message":"Manages build reqs in alexandria project\n\nChange-Id: Idf25ce004619a0e6d30bce0522236c0b4d2537f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0540284625bd678fa244d24f86ad151a42c58bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0540284625bd678fa244d24f86ad151a42c58bc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
