)]}'
{"id":"openstack%2Fproject-config~679743","triplet_id":"openstack%2Fproject-config~master~I304852bbeb1b82140690ccdb9c011ea0b1c02c28","project":"openstack/project-config","branch":"master","topic":"networking-generic-switch-tempest-plugin","hashtags":[],"change_id":"I304852bbeb1b82140690ccdb9c011ea0b1c02c28","subject":"Finish retiring networking-generic-switch-tempest-plugin","status":"MERGED","created":"2019-09-03 10:29:49.000000000","updated":"2019-09-05 06:48:32.000000000","submitted":"2019-09-05 06:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679743-1567666112507-673e73e3","meta_rev_id":"79bcfd87570c887af182c20f6a3b799c24ed87f8","_number":679743,"virtual_id_number":679743,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-05 06:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 16:03:00.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-05 06:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-03 16:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-05 06:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 16:08:06.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-05 06:35:49.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-05 06:48:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc42cb03ac07b8fcbbd13d9ba06399987f40615f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-03 10:29:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62437f50e711d6f9f3351eeea705f4b0e6292ae4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 10:43:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2ca699917ef14796a8c6a8dd6478e0e5 : SUCCESS in 6m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5f9fe0d550174ee48578e0e5c9649e24 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9b302ce8e357a55537b7a32ed84ec909fee485bc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 11:04:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b4219dd485e53b000d537dbd97ee45d46bb2dafc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 11:05:38.000000000","message":"Patch Set 1:\n\nyou need a change for governance repo as well, see https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L1289","accounts_in_message":[],"_revision_number":1},{"id":"2f403909d720300101252e6172409ad16074ac60","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-03 11:10:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02a7498706c099346fe5f3fdef77982941fb55ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:19:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/86f39c61782e4ab0848302563c2dd51d : SUCCESS in 6m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cd8000d5aa454e559bcde6dc4fee1e42 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"bdf00481eead1f059347405f70c5215967d56c3e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 15:56:37.000000000","message":"Patch Set 2:\n\nLGTM, waiting for change to governance before I +2.","accounts_in_message":[],"_revision_number":2},{"id":"7b1b183d6012786c2a7e27f1d8b6d4867f929992","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-03 16:03:00.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/#/c/679816/","accounts_in_message":[],"_revision_number":2},{"id":"e6fc2ce96ef5e575b3f24a192b8f85b4b725c42a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 16:08:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c73b8770d307f293ec8bd3b8cce8ae8e164b2c5d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-05 06:35:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8e2b286989cfa3a943310045aa0cf44f26ded43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:36:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bcc3de34fdd613737d33fbe45318d536fa2191be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:48:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d1bb03aef8f86ce5e6bc5d05381efa8bdcc31fbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:48:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b2078063fef54f728c72bacd0abea249 : SUCCESS in 6m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/da6577c3f252493a8e9da8c1b545f8b6 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"133be5c2d30bdd31a670177f622b1145ef0286cf","revisions":{"161ed616b4a3895c90bccb8cb20c3cca686b739d":{"kind":"REWORK","_number":1,"created":"2019-09-03 10:29:49.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/43/679743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/679743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/679743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/679743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/679743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/679743/1"}}},"commit":{"parents":[{"commit":"05e2bdba103bd9ca912925761f231cdc91c3bd18","subject":"Lowering max-servers for FN to accomdate NUMA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05e2bdba103bd9ca912925761f231cdc91c3bd18"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-03 10:29:12.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-03 10:29:31.000000000","tz":120},"subject":"Finish retiring networking-generic-switch-tempest-plugin","message":"Finish retiring networking-generic-switch-tempest-plugin\n\nThis project is being retired as it\u0027s not a real tempest plugin and\nshould not be shipped or used outside of networking-generic-switch CI.\n\nChange-Id: I304852bbeb1b82140690ccdb9c011ea0b1c02c28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/161ed616b4a3895c90bccb8cb20c3cca686b739d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/161ed616b4a3895c90bccb8cb20c3cca686b739d"}]},"branch":"refs/heads/master"},"133be5c2d30bdd31a670177f622b1145ef0286cf":{"kind":"REWORK","_number":2,"created":"2019-09-03 11:10:31.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/43/679743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/679743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/679743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/679743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/679743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/679743/2"}}},"commit":{"parents":[{"commit":"05e2bdba103bd9ca912925761f231cdc91c3bd18","subject":"Lowering max-servers for FN to accomdate NUMA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05e2bdba103bd9ca912925761f231cdc91c3bd18"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-03 10:29:12.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-03 11:10:22.000000000","tz":120},"subject":"Finish retiring networking-generic-switch-tempest-plugin","message":"Finish retiring networking-generic-switch-tempest-plugin\n\nThis project is being retired as it\u0027s not a real tempest plugin and\nshould not be shipped or used outside of networking-generic-switch CI.\n\nChange-Id: I304852bbeb1b82140690ccdb9c011ea0b1c02c28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/133be5c2d30bdd31a670177f622b1145ef0286cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/133be5c2d30bdd31a670177f622b1145ef0286cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
