)]}'
{"id":"openstack%2Fproject-config~679789","triplet_id":"openstack%2Fproject-config~master~Ib814d1b33132e1846b74295b30979ae3d5eafd36","project":"openstack/project-config","branch":"master","topic":"674091","hashtags":[],"change_id":"Ib814d1b33132e1846b74295b30979ae3d5eafd36","subject":"Adding in proper NUMA labels for FN","status":"MERGED","created":"2019-09-03 14:10:12.000000000","updated":"2019-09-03 16:10:25.000000000","submitted":"2019-09-03 16:10:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679789-1567527025233-bbbf3676","meta_rev_id":"f239e49a1d8fec07d8432fe0dd9a3f14c6315d32","_number":679789,"virtual_id_number":679789,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-03 15:21:22.000000000","_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-03 16:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":2,"date":"2019-09-03 15:24:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-03 15:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-09-03 15:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 15:24:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-03 15:55:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-03 16:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a11e97dfa64bad5e4a85b5229227e392254c6f8f","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-09-03 14:10:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e0de355b46c35974f673f6dafdc6e481f2ef665","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 14:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/053bd7efff1a4b91b5ec27e02d324902 : SUCCESS in 6m 39s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/590f379ebd8d48c58dc7686fa3d65789 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a7c38876be1e13a5857124c8fce87d5aa5115318","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-03 15:15:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66a1c1c39e84877971d66ab98acbe08450736220","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-09-03 15:20:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba943d42764a852bb9ca56de631a7b6187a30b29","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-09-03 15:21:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nOk, does this look a little better?","accounts_in_message":[],"_revision_number":2},{"id":"b9116b28643af32f65a40d2dc60aaf9ae7b881ec","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-03 15:24:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c001496170780e49f245d9ecef719ba84ae2dfce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 15:30:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ec7dc0c243834125b8a6fa09fcac94af : SUCCESS in 7m 37s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c8ce61b363814c2ebfe5cbd81cf7c372 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"07659de66621970ae75dac02ccae8fc8d364e415","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 15:55:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"31985828bda507ef84fb2ee50f5b33ddf569e7c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 15:55:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f72aed5880ce802111ae9c7989a6b48ff8614380","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 16:10:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eeb498fa41344430abe421e0c89bc6e5 : SUCCESS in 8m 19s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/5acf48d6bac745a5ad051643db3af514 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"76ff38da23675f2f8641fb19b3917e98d6162924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 16:10:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"747a1283e82ff60e2d652b085909b78477c4f8a7","revisions":{"1bd4e69f9d8fe12f53396afdbcd9cb4859121c41":{"kind":"REWORK","_number":1,"created":"2019-09-03 14:10:12.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/89/679789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/679789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/679789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/679789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/679789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/679789/1"}}},"commit":{"parents":[{"commit":"05e2bdba103bd9ca912925761f231cdc91c3bd18","subject":"Lowering max-servers for FN to accomdate NUMA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05e2bdba103bd9ca912925761f231cdc91c3bd18"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-09-03 14:09:34.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-09-03 14:09:34.000000000","tz":-240},"subject":"Adding in proper NUMA labels for FN","message":"Adding in proper NUMA labels for FN\n\nChange-Id: Ib814d1b33132e1846b74295b30979ae3d5eafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bd4e69f9d8fe12f53396afdbcd9cb4859121c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bd4e69f9d8fe12f53396afdbcd9cb4859121c41"}]},"branch":"refs/heads/master"},"747a1283e82ff60e2d652b085909b78477c4f8a7":{"kind":"REWORK","_number":2,"created":"2019-09-03 15:20:22.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/89/679789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/679789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/679789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/679789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/679789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/679789/2"}}},"commit":{"parents":[{"commit":"05e2bdba103bd9ca912925761f231cdc91c3bd18","subject":"Lowering max-servers for FN to accomdate NUMA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05e2bdba103bd9ca912925761f231cdc91c3bd18"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-09-03 14:09:34.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-09-03 15:20:17.000000000","tz":-240},"subject":"Adding in proper NUMA labels for FN","message":"Adding in proper NUMA labels for FN\n\nChange-Id: Ib814d1b33132e1846b74295b30979ae3d5eafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
