)]}'
{"id":"openstack%2Fproject-config~679848","triplet_id":"openstack%2Fproject-config~master~I7fc56e89ab3dff2c3106814b64b370d227e7d0ea","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I7fc56e89ab3dff2c3106814b64b370d227e7d0ea","subject":"StarlingX retructuring: Ten new repos","status":"MERGED","created":"2019-09-03 19:24:07.000000000","updated":"2019-09-04 19:54:54.000000000","submitted":"2019-09-04 19:54:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679848-1567626894112-36715ff1","meta_rev_id":"fef252c067f4ee091ba820edd52165a22c8fbbf0","_number":679848,"virtual_id_number":679848,"owner":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":2,"date":"2019-09-04 19:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-04 19:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-04 18:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":1,"date":"2019-09-04 19:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-09-04 19:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 19:35:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-04 19:39:57.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-09-04 19:40:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-04 19:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d1fc5b4363566d3c9eb3b9593e8d67745e1934a","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-03 19:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e48169119e157ba9b22c4dd0bdd9608c91f8d6f6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 19:30:35.000000000","message":"Patch Set 1:\n\nI suggest you WIP this for now...","accounts_in_message":[],"_revision_number":1},{"id":"6c373292211bdb00e85cce95ec659947923400cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:32:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/37402326940a4b588e0edf29749c05d0 : SUCCESS in 6m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c045abba1792485d8a286f8485d53131 : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"087ec2631c5e5c4fdcca63c8dd33b78a54917dd4","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-03 19:52:22.000000000","message":"Patch Set 1: Workflow-1\n\nWork in Progress","accounts_in_message":[],"_revision_number":1},{"id":"a8038b4d6d8bbb5329b580ab53ec83d142fddb2c","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-03 20:14:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e7983215e554484d391b1fac6505b01f54e8f176","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:29:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c35a4c089cc54237ad834ebe9d8c655e : SUCCESS in 7m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/377f0d2e914a4ba99b0ceb57c4737323 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c769cfc80ebb4e43e8b6eb83d1b39cf04c240ca2","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-04 15:26:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1cb2d199e20f56095a734c02f11b9b2e84da5d3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 15:54:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4330ded1cb1547288d610326be5164e4 : SUCCESS in 7m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1a6cd93aabcc4b78bba8c38589edf78b : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a639475abcec70a87af590b51e5cc30ffff2f0b5","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-04 15:59:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9670239b73c70fbebb30afb7bfef685ec9071c14","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-09-04 16:03:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2bc29229be89a7d1b0bfe80eabf89e940fcc5b9c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 16:04:46.000000000","message":"Patch Set 4: Code-Review+2\n\nrepos looked fine in previous run, this should be good.","accounts_in_message":[],"_revision_number":4},{"id":"dd1890449c3d05145ea3a71d1fd63c05638705ed","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-04 16:11:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d59740d1d2096638d83ce5e9091dc6d47122c4e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 16:54:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e8987d5379c44741b4c9b45444d7caea : SUCCESS in 8m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b19b7874f7b04316b32ed21b2ce3fa8b : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"24790449cbc43822b74f72776bedf2d825f47c2e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-04 18:25:53.000000000","message":"Patch Set 4: Code-Review+1\n\nThis change looks fine to me. Just had a small question on irc about whether or not we want \"kubernetes\" and \"puppet\" repos as they may appear to be forks at first glance. Other than that I would approve as is.","accounts_in_message":[],"_revision_number":4},{"id":"ba8cfe6b26a24ea60b89d5d61432a36c39498839","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-04 18:46:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f00ea4578c25d02cc4a3ebd02d2987eafce0e5e6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 18:48:52.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b3af363723780784a2ae79322dec1fe4c9671bfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:59:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4cb9c2f3a54941d1a40cbecbb85288ee : SUCCESS in 8m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4d48d81b6fef451a82ddc3b9803bedf3 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":5},{"id":"8ff20ddf320fbb3e62ebb454ed6f3e2b1dd56b1e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-04 19:35:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2d5781e3334c890fcc474659e93e485d71e8c05e","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-09-04 19:39:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"600e843f74fe3e5d5acc170e738a43ccf9408783","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 19:40:57.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"91df19f7ab163c9b40a5e98e839cdd00aa63e1cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 19:41:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9da1cf4b59274480b3377ddb1ead941542750f0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 19:54:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5660c060454145a1af59af17fef2f846 : SUCCESS in 8m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/645eaace06cf4a35977056c136c62cf4 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":5},{"id":"fef252c067f4ee091ba820edd52165a22c8fbbf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 19:54:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e0b0824ecfcd7c4dce179d9ebef9f24846d56af1","revisions":{"be1fb80e6a004221546367f407f59f25b70f49e4":{"kind":"REWORK","_number":1,"created":"2019-09-03 19:24:07.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/48/679848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/679848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/679848/1"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 19:21:41.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 19:22:27.000000000","tz":-240},"subject":"StarlingX retructuring: Ten new repos","message":"StarlingX retructuring: Ten new repos\n\nChange-Id: I7fc56e89ab3dff2c3106814b64b370d227e7d0ea\nStory: 2006166\nTask: 35687\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be1fb80e6a004221546367f407f59f25b70f49e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be1fb80e6a004221546367f407f59f25b70f49e4"}]},"branch":"refs/heads/master"},"4eb6640a1b47f82da9afafcf4d9e155713cae9f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-03 20:14:08.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/48/679848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/679848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/679848/2"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 19:21:41.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 20:06:13.000000000","tz":-240},"subject":"StarlingX retructuring: Ten new repos","message":"StarlingX retructuring: Ten new repos\n\nStarlings is spliting up several of it\u0027s existing\nrepos.\n\nThe goals are:\n\n1) Allow the build to be preformed in logical layers.\nThis will allow faster build times for most developers\nwho are uninterested in e.g. the distro layer.\n\n2) Isolate StarlingX original code from patches\nagains third party code.\n\n3) Isolate armada applictions from one another.\n\n4) Seperate high churn config files from low churn apps.\n\nChange-Id: I7fc56e89ab3dff2c3106814b64b370d227e7d0ea\nStory: 2006166\nTask: 35687\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4eb6640a1b47f82da9afafcf4d9e155713cae9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4eb6640a1b47f82da9afafcf4d9e155713cae9f1"}]},"branch":"refs/heads/master"},"158e3bde2f42cb08592da6c413e5fba79989ed43":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-04 15:26:05.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/48/679848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/679848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/679848/3"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 19:21:41.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-04 15:24:55.000000000","tz":-240},"subject":"StarlingX retructuring: Ten new repos","message":"StarlingX retructuring: Ten new repos\n\nStarlings is spliting up several of it\u0027s existing\nrepos.\n\nThe goals are:\n\n1) Allow the build to be preformed in logical layers.\nThis will allow faster build times for most developers\nwho are uninterested in e.g. the distro layer.\n\n2) Isolate StarlingX original code from patches\nagains third party code.\n\n3) Isolate armada applictions from one another.\n\n4) Seperate high churn config files from low churn apps.\n\nNew repos created:\n\nstarlingx/compile\nstarlingx/config-files\nstarlingx/helm-charts\nstarlingx/kubernetes\nstarlingx/monitor-armada-app\nstarlingx/monitoring\nstarlingx/openstack-armada-app\nstarlingx/platform-armada-app\nstarlingx/puppet\nstarlingx/utilities\n\nChange-Id: I7fc56e89ab3dff2c3106814b64b370d227e7d0ea\nStory: 2006166\nTask: 35687\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/158e3bde2f42cb08592da6c413e5fba79989ed43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/158e3bde2f42cb08592da6c413e5fba79989ed43"}]},"branch":"refs/heads/master"},"5de3c21575e726e5fcbb1542d79977061db995a6":{"kind":"REWORK","_number":4,"created":"2019-09-04 15:59:45.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/48/679848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/679848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/679848/4"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 19:21:41.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-04 15:58:16.000000000","tz":-240},"subject":"StarlingX retructuring: Ten new repos","message":"StarlingX retructuring: Ten new repos\n\nStarlings is spliting up several of it\u0027s existing\nrepos.\n\nThe goals are:\n\n1) Allow the build to be preformed in logical layers.\nThis will allow faster build times for most developers\nwho are uninterested in e.g. the distro layer.\n\n2) Isolate StarlingX original code from patches\nagains third party code.\n\n3) Isolate armada applictions from one another.\n\n4) Seperate high churn config files from low churn apps.\n\nNew repos created:\n\nstarlingx/compile\nstarlingx/config-files\nstarlingx/helm-charts\nstarlingx/kubernetes\nstarlingx/monitor-armada-app\nstarlingx/monitoring\nstarlingx/openstack-armada-app\nstarlingx/platform-armada-app\nstarlingx/puppet\nstarlingx/utilities\n\nChange-Id: I7fc56e89ab3dff2c3106814b64b370d227e7d0ea\nStory: 2006166\nTask: 35687\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5de3c21575e726e5fcbb1542d79977061db995a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5de3c21575e726e5fcbb1542d79977061db995a6"}]},"branch":"refs/heads/master"},"e0b0824ecfcd7c4dce179d9ebef9f24846d56af1":{"kind":"REWORK","_number":5,"created":"2019-09-04 18:46:33.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/48/679848/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/679848/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/679848/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/679848/5"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-03 19:21:41.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2019-09-04 18:45:47.000000000","tz":-240},"subject":"StarlingX retructuring: Ten new repos","message":"StarlingX retructuring: Ten new repos\n\nStarlings is spliting up several of it\u0027s existing\nrepos.\n\nThe goals are:\n\n1) Allow the build to be preformed in logical layers.\nThis will allow faster build times for most developers\nwho are uninterested in e.g. the distro layer.\n\n2) Isolate StarlingX original code from patches\nagains third party code.\n\n3) Isolate armada applictions from one another.\n\n4) Seperate high churn config files from low churn apps.\n\nNew repos created:\n\nstarlingx/compile\nstarlingx/config-files\nstarlingx/containers\nstarlingx/helm-charts\nstarlingx/monitor-armada-app\nstarlingx/monitoring\nstarlingx/openstack-armada-app\nstarlingx/platform-armada-app\nstarlingx/stc-puppet\nstarlingx/utilities\n\nChange-Id: I7fc56e89ab3dff2c3106814b64b370d227e7d0ea\nStory: 2006166\nTask: 35687\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0b0824ecfcd7c4dce179d9ebef9f24846d56af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0b0824ecfcd7c4dce179d9ebef9f24846d56af1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
