)]}'
{"id":"openstack%2Fproject-config~679850","triplet_id":"openstack%2Fproject-config~master~I3757ca85b64b12c90741a8d5075f59ddb65d549f","project":"openstack/project-config","branch":"master","topic":"starlingx-rename-irc","hashtags":[],"change_id":"I3757ca85b64b12c90741a8d5075f59ddb65d549f","subject":"Remove broken notification for starlingx","status":"MERGED","created":"2019-09-03 19:30:25.000000000","updated":"2019-09-05 06:46:06.000000000","submitted":"2019-09-05 06:46:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679850-1567665966197-0552f13f","meta_rev_id":"84a9e9ea0f3d0ecd57278df5808ad26b6b2e2154","_number":679850,"virtual_id_number":679850,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2019-09-04 17:09:58.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":2,"date":"2019-09-05 06:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-05 06:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-04 16:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":1,"date":"2019-09-04 16:11:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-05 06:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 16:11:34.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-09-04 16:11:58.000000000","updated_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2019-09-05 06:33:41.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-05 06:46:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"705872e310f70f097fcd95a579ccbd24d3446a3c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-03 19:30:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98c2b5c919673f2fe600f773c697671d0a6b5759","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:49:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eb909e9b545548f49b5e3c1f202af263 : SUCCESS in 8m 42s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b940b9c411ce4319967075e6a2a77e11 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5a78e52d7ac97dc64b57aba974c8dacfd9390295","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 16:10:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0361bb3cabe0945b4b6a96399d05593eb52f205a","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-09-04 16:11:34.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Andreas","accounts_in_message":[],"_revision_number":2},{"id":"92d609ba29b2b7709445b2dea556cfb933871179","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2019-09-04 16:11:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0a9b6b42b72e0c1e396f3f7f0eeb92756be36b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 16:24:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/58781eebcb5b4cd581c438c234856a41 : SUCCESS in 6m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/17c6ba80ec044ae6a3c90d5704f760ba : POST_FAILURE in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"de2000c809d5844e6ae6e56f48f168d3bb0ca6f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-04 17:09:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ddc73b9a4e125384d041b29ceb88ef2d9ea2dd49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:25:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5418ffcbac834d8d923a4d445df24266 : SUCCESS in 6m 53s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ad72013a0d924da680410c4ed92a032a : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":2},{"id":"dc18f37d2454298a9783c5c3185e2a2800759b9e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-05 06:33:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b5ab8ea1dcd11387bf98b5b98460ed69382f54c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:33:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9442717b04fefcf23ed4779b2a08e9d901b5dc9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:46:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3cb58cd19e07414599595c35e1d75e97 : SUCCESS in 6m 25s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/75e69d40989a4b12bcb957d2cc30bce9 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5cd9c2d8fd3d071db24fdc1431f1f17dc476c020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:46:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04de404dcedd8a1f5a6d710e280ac90859875625","revisions":{"1e0e8ef609b523efeb6232789e4bc1c701f8ed25":{"kind":"REWORK","_number":1,"created":"2019-09-03 19:30:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/679850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/679850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/679850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/679850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/679850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/679850/1"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-03 19:29:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-03 19:29:29.000000000","tz":120},"subject":"Update channels for starlingx","message":"Update channels for starlingx\n\nThe OpenDev rename missed the stx repos due to the regex, update so\nthat starlingx gets IRC notifications again.\n\nChange-Id: I3757ca85b64b12c90741a8d5075f59ddb65d549f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e0e8ef609b523efeb6232789e4bc1c701f8ed25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e0e8ef609b523efeb6232789e4bc1c701f8ed25"}]},"branch":"refs/heads/master"},"04de404dcedd8a1f5a6d710e280ac90859875625":{"kind":"REWORK","_number":2,"created":"2019-09-04 16:10:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/679850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/679850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/679850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/679850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/679850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/679850/2"}}},"commit":{"parents":[{"commit":"747a1283e82ff60e2d652b085909b78477c4f8a7","subject":"Adding in proper NUMA labels for FN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/747a1283e82ff60e2d652b085909b78477c4f8a7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-03 19:29:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-04 16:09:39.000000000","tz":120},"subject":"Remove broken notification for starlingx","message":"Remove broken notification for starlingx\n\nThe OpenDev rename missed the stx repos due to the regex, and\nfeedback by starlingx folks that this is nice - the notifications were\ntoo much noise.\n\nSo, remove the broken notifications.\n\nChange-Id: I3757ca85b64b12c90741a8d5075f59ddb65d549f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04de404dcedd8a1f5a6d710e280ac90859875625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04de404dcedd8a1f5a6d710e280ac90859875625"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
