)]}'
{"id":"openstack%2Fproject-config~680298","triplet_id":"openstack%2Fproject-config~master~I9b0ee60f87a20a4ab249b8f0cf469a205217b504","project":"openstack/project-config","branch":"master","topic":"project-yaml-normalization","hashtags":[],"change_id":"I9b0ee60f87a20a4ab249b8f0cf469a205217b504","subject":"Normalize projects.yaml","status":"MERGED","created":"2019-09-05 06:16:04.000000000","updated":"2019-09-05 07:38:21.000000000","submitted":"2019-09-05 07:38:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680298-1567669101422-233c4fa7","meta_rev_id":"09e71be76ffac6deedbffee6f9563e7d856882d1","_number":680298,"virtual_id_number":680298,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-05 07:38:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-09-05 06:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-09-05 06:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 06:21:08.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-05 07:38:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5357a94069b4633429e10e68cc6dbdcce545c83e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-05 06:16:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b69e26345de075f6bcf53c3049c002dd262558b1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 06:21:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrepos are imported","accounts_in_message":[],"_revision_number":1},{"id":"118fcd5939b330086bcfc70e982d8b082a9fed9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 07:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a8403520c2247ecb323658d60a81f42 : SUCCESS in 7m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5e7c39d7c1a642a8ae534657030671f6 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"92c812298dc7d0e2eeb5e54368e8c6588c547acf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 07:21:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"892d935f93d568add0cc3dfe7ab69620b64dd9dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 07:38:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a96802b74b003df905e2c9f5864fc16ff731f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 07:38:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/668d49a988164149a1fe6751c2ccce8f : SUCCESS in 10m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9a59cca0948a45bb99c0b02be29c8e65 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ca470e532ca82e56c2bf049ab20a8e9e64639e5","revisions":{"8ca470e532ca82e56c2bf049ab20a8e9e64639e5":{"kind":"REWORK","_number":1,"created":"2019-09-05 06:16:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/680298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/98/680298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/680298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/680298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/680298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/98/680298/1"}}},"commit":{"parents":[{"commit":"e0b0824ecfcd7c4dce179d9ebef9f24846d56af1","subject":"StarlingX retructuring: Ten new repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0b0824ecfcd7c4dce179d9ebef9f24846d56af1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 06:16:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 06:16:03.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I9b0ee60f87a20a4ab249b8f0cf469a205217b504\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ca470e532ca82e56c2bf049ab20a8e9e64639e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ca470e532ca82e56c2bf049ab20a8e9e64639e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
