)]}'
{"id":"openstack%2Fproject-config~680805","triplet_id":"openstack%2Fproject-config~master~Ief6df503a0dfaee4b988db0461d52c0eed3ae7fa","project":"openstack/project-config","branch":"master","topic":"674091","hashtags":[],"change_id":"Ief6df503a0dfaee4b988db0461d52c0eed3ae7fa","subject":"Scaling FN down to 0 for backup gen install","status":"MERGED","created":"2019-09-07 13:32:49.000000000","updated":"2019-09-07 15:48:27.000000000","submitted":"2019-09-07 15:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680805-1567871307817-72f0986f","meta_rev_id":"0a943e77bfcdc350ef8182f02a5c356366554f7c","_number":680805,"virtual_id_number":680805,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":2,"date":"2019-09-07 15:48:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"all":[{"value":1,"date":"2019-09-07 15:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-07 13:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-07 13:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 13:34:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-07 15:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b78dfd0826a247ef5a4e0e93cbca597690043ddc","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-09-07 13:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1eefe9eff911e78d20ffd0e54049530929f8dba","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-07 13:34:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5947d246124ad3b631ce2d396c5b437c8e1eab91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 13:43:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e3694d57a64c4b6aab3196b9ab8b7b88 : SUCCESS in 5m 38s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/a9d47cc5f793465499eef67826a1df9a : POST_FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"83c0697d8211ef0d95f0a91a9fce30721ec81d10","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-07 14:10:39.000000000","message":"Patch Set 1:\n\nrecheck due to POST_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5712e17557e07751f8e92ebb3c5d34fe50800f56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 14:22:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1c00ee4b6a4e4ddaae5bac433e6e2332 : POST_FAILURE in 6m 11s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/59d597b17e7e444eb19a61f7426e35ef : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f86f1bf2b6c561afb51a55927169c4f274c6aaab","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-07 14:25:30.000000000","message":"Patch Set 1:\n\nrecheck again due to POST_FAILURE now that FN has been taken out of the log storage rotation","accounts_in_message":[],"_revision_number":1},{"id":"747e719c1cf4baec7f5ba6f28c97c29db63dc3ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 14:38:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be9efb053ea24a3b91361400636f2420 : SUCCESS in 6m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3d5a97ab687041b59776cf7e75d4be8f : POST_FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"98398276a4cc949a5d2b1371ff5aad30c57e5840","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-09-07 15:09:43.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck now that ovh is out","accounts_in_message":[],"_revision_number":1},{"id":"084645f3dba6ca851ed81fc29baa58ebcbf3d4e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 15:21:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2c89b3e5262948a4b7f6e39b598461f8 : SUCCESS in 6m 21s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/a4dda6dc76ff4298b30eea82701f6f87 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"53c05ebb4ec367f2c1194e5680d1908a90153d38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 15:22:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7eb6a1b6e7b7e3f39572bc25ad17ac4a822cca3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 15:48:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9756a398cc62e4323f3a8bd5a0ba45a381eecbe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 15:48:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b9179283daed47ff8d3cf6e2160ee0be : SUCCESS in 6m 00s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/fec761ab79d74356974c09cd903e6d0b : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19d1c9650d161a5b1136fb82b5af7c67412e5841","revisions":{"19d1c9650d161a5b1136fb82b5af7c67412e5841":{"kind":"REWORK","_number":1,"created":"2019-09-07 13:32:49.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/05/680805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/680805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/680805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/680805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/680805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/680805/1"}}},"commit":{"parents":[{"commit":"e1c2c349260566ae8db604a55c3f819d0e236ea0","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1c2c349260566ae8db604a55c3f819d0e236ea0"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-09-07 13:32:26.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2019-09-07 13:32:26.000000000","tz":-240},"subject":"Scaling FN down to 0 for backup gen install","message":"Scaling FN down to 0 for backup gen install\n\nChange-Id: Ief6df503a0dfaee4b988db0461d52c0eed3ae7fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/19d1c9650d161a5b1136fb82b5af7c67412e5841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/19d1c9650d161a5b1136fb82b5af7c67412e5841"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
