)]}'
{"id":"openstack%2Fproject-config~680830","triplet_id":"openstack%2Fproject-config~master~I97fc69c63b92c27dd33abc35d4e0fc7b72d74eda","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I97fc69c63b92c27dd33abc35d4e0fc7b72d74eda","subject":"Add windmill-ops / ansible-role-elasticsearch","status":"MERGED","created":"2019-09-07 18:39:55.000000000","updated":"2019-09-18 18:18:58.000000000","submitted":"2019-09-18 18:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680830-1568830738457-fa3c3f25","meta_rev_id":"9fcfe52e7622b79db1a67382076c2356f979f422","_number":680830,"virtual_id_number":680830,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-18 18:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-18 18:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-18 17:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-18 18:04:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 17:40:50.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-18 18:04:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-18 18:18:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e42b814c4d702eb03cc5e064c982743e980d6fa4","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-09-07 18:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c9d8a18dd5e01ff0aff9785c196a6f0400ec3d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 18:53:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/abe4601ef1754629a4ca38166df095ca : SUCCESS in 7m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c83ab90633e4463eaf5710220e5b96fd : SUCCESS in 3m 14s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b19f9443d0494cffb3d5cd0de092a748 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d8cb5eb858a9d97b4a2104b2b03cccfca3c06d5b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-08 13:02:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c766d144bbce8bb2e1be73dbf770f4df5e51ff2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-18 17:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9818ad23fdd0c3e92f56c64d22fd4ef74f39e0d1","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-09-18 17:40:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94026d8a794c715fd76fc82213219972fbeef55e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 17:40:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5cedaf6c51b14255e1479eadaece9aeb2033f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:53:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64b99ddf6b074dc3a33b64285c023746 : SUCCESS in 5m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c61c624fa43d4a68aaf566085fd420aa : SUCCESS in 3m 22s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/500da7467bcc4d6187f81c623f4261a9 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e8ded6a2efa18b536724bdcc0be328133dc01227","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-18 18:04:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2776d41e64e06bc02ba60664a1e5df2bcda3170b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:04:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3907432c970f9811120690db85e9fb86aa0cd173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:18:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4c36d742a3c54999babc1e8b6caf197b : SUCCESS in 6m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fa159ee64d7c47d7b0397b93d225f45b : SUCCESS in 4m 45s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/d2519d4faa8045539e696faf6c2a0f59 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9fcfe52e7622b79db1a67382076c2356f979f422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:18:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e44900bf6eb124dd466f8c142f6d2244fd27373","revisions":{"d55501e517366888135d3771c7695aa6dd3b84bb":{"kind":"REWORK","_number":1,"created":"2019-09-07 18:39:55.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/30/680830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/680830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/680830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/680830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/680830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/680830/1"}}},"commit":{"parents":[{"commit":"19d1c9650d161a5b1136fb82b5af7c67412e5841","subject":"Scaling FN down to 0 for backup gen install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/19d1c9650d161a5b1136fb82b5af7c67412e5841"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-09-07 18:38:30.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-09-07 18:38:30.000000000","tz":-240},"subject":"Add windmill-ops / ansible-role-elasticsearch","message":"Add windmill-ops / ansible-role-elasticsearch\n\nThese repos will be used to start adding elastic search support to\nwindmill. Day 2 operations related to digging deeper into zuul logs.\n\nChange-Id: I97fc69c63b92c27dd33abc35d4e0fc7b72d74eda\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d55501e517366888135d3771c7695aa6dd3b84bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d55501e517366888135d3771c7695aa6dd3b84bb"}]},"branch":"refs/heads/master"},"4e44900bf6eb124dd466f8c142f6d2244fd27373":{"kind":"REWORK","_number":2,"created":"2019-09-18 17:40:07.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/30/680830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/680830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/680830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/680830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/680830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/680830/2"}}},"commit":{"parents":[{"commit":"19d1c9650d161a5b1136fb82b5af7c67412e5841","subject":"Scaling FN down to 0 for backup gen install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/19d1c9650d161a5b1136fb82b5af7c67412e5841"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-09-07 18:38:30.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-09-18 17:39:50.000000000","tz":-240},"subject":"Add windmill-ops / ansible-role-elasticsearch","message":"Add windmill-ops / ansible-role-elasticsearch\n\nThese repos will be used to start adding elastic search support to\nwindmill. Day 2 operations related to digging deeper into zuul logs.\n\nChange-Id: I97fc69c63b92c27dd33abc35d4e0fc7b72d74eda\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e44900bf6eb124dd466f8c142f6d2244fd27373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e44900bf6eb124dd466f8c142f6d2244fd27373"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
