)]}'
{"id":"openstack%2Fproject-config~681259","triplet_id":"openstack%2Fproject-config~master~I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10","project":"openstack/project-config","branch":"master","topic":"retire-charm-neutron-api-genericswitch","hashtags":[],"change_id":"I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10","subject":"End project gating for charm-neutron-api-genericswitch","status":"MERGED","created":"2019-09-10 13:47:00.000000000","updated":"2019-09-13 20:30:43.000000000","submitted":"2019-09-13 20:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681259-1568406643346-9b7968ff","meta_rev_id":"bc771135c5a88c2d1dbef449c5865f33209fd18b","_number":681259,"virtual_id_number":681259,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-13 20:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-13 20:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-10 16:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-13 20:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 16:28:32.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-13 20:19:32.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-13 20:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0a4351f5daccc11d12671c14f652a394b261996","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 13:47:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b51c24f58b48ae4d77fae4c7bd90800ada234ba7","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 13:54:39.000000000","message":"Topic set to retire-charm-neutron-api-genericswitch","accounts_in_message":[],"_revision_number":1},{"id":"72dc0386c001c37f63a830bd9a5dec5a15860b95","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 13:57:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fa17a3e7b74091ebc6aaba79aa5c0b9de2873bf","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:03:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ccb773573042ed53b7a439f5bbb0dc55e2526cc3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 14:03:15.000000000","message":"Patch Set 2: Code-Review-1\n\nRetiring a repo is a multiple step approach, follow all steps from:\n\nhttps://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nPlease also add the link to your announcement to the commit message - and push up all changes together via needed-by/depends-on.\n\nI need to see:\n* two changes for this repo\n* one change for governance\n* one change for the to be retired repo to \"empty\" it","accounts_in_message":[],"_revision_number":2},{"id":"2b446623a80becff14ba9e316b39839245fedf13","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 14:03:36.000000000","message":"Patch Set 3: Code-Review-1\n\nsee previous comment","accounts_in_message":[],"_revision_number":3},{"id":"992256414d5f22250c8ac0d25b926aa72dd09a7a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 14:04:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c087cdc19c2aac7dd99db179ebc1daf34d67ffd","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:10:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"839f4fa140daf078254ceee3e8c30a9da90ca54a","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:14:39.000000000","message":"Topic changed from end-gating to charm-neutron-api-genericswitch","accounts_in_message":[],"_revision_number":4},{"id":"e52fae126202bdf3166a93446fbf45dd8fd64ee8","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:14:54.000000000","message":"Topic changed from charm-neutron-api-genericswitch to retire-charm-neutron-api-genericswitch","accounts_in_message":[],"_revision_number":4},{"id":"f386ce9515fc568453cb3dba80bcadbf3f58f512","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 14:16:53.000000000","message":"Patch Set 4: Code-Review-1\n\nwrong file, see previous comment","accounts_in_message":[],"_revision_number":4},{"id":"301d07a6d0fa9104289351fc52e86746f0d62005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:23:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b957a27550dd4e5197b9943b0154401b : SUCCESS in 6m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fe55b75247bc49e8bf8cc8d3576b862e : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"939e204cb2c295c059729af684456cbc75afd91b","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:34:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db32e5d61504ac6fbba81c6a284032495e4f3c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:34:33.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).","accounts_in_message":[],"_revision_number":5},{"id":"d006f400d629ba8f1c7130fa9cf31f5d5be3eec0","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:35:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"602af9dd050c60f646f8b7f2358244357be764c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:36:03.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).","accounts_in_message":[],"_revision_number":6},{"id":"16d42d3dcb6b260b3b6d32530c1a131a9347e2f1","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:37:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7bac044c3c5525dbe6a04ed63269b6ae43475beb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 14:38:07.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1a44295be1f3421d01e326297dee741d162cc020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:42:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2260c99078b94e30acc75edae9e6bd10 : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":7},{"id":"45f7a3b1f170f35ab374fb9603412f500387f0d8","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:43:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0465b50b03fec471e5dbc5ec11de62f1c984ca0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:52:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0e5e3d7869224dfbb703a2a49946cae1 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":8},{"id":"8708c1146a1c4d0ba2307bc35cd90c9d8a24c4c0","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 15:14:24.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"7b996b0e2ccf6703dde38c35e6ba3fb3f6310dbc","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 15:15:58.000000000","message":"Topic changed from remove-charm-neutron-api-genericswitch to retire-charm-neutron-api-genericswitch","accounts_in_message":[],"_revision_number":9},{"id":"39869ecd5bf3cc2bcb8e32a859cdc33124931475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:23:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cdfedeaff0634fb5a3374d293d1229b4 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":9},{"id":"3f084c8fff0099e377f9893cf2aa370449105d17","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 16:28:32.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"1ba210ca6c87498fc0b50e9690204a8387aad41b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-13 20:19:32.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"2d7d6e750ccd17442dd906f5d1182f36ff669661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 20:19:52.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"43e3e59d6b9c056e2003e886b3ef6a000fffc25c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 20:30:42.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f354909f913541f086ecdcb0f3509732 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":9},{"id":"bc771135c5a88c2d1dbef449c5865f33209fd18b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 20:30:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"995ef5c0b35d5c33a629be2ea32692cd9bf881ab","revisions":{"51149aa7fb84263ea7eef2ce100f24905c444a2c":{"kind":"REWORK","_number":1,"created":"2019-09-10 13:47:00.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/1"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:59.000000000","tz":-240},"subject":"Retire charm-neutron-api-genericswitch","message":"Retire charm-neutron-api-genericswitch\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51149aa7fb84263ea7eef2ce100f24905c444a2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51149aa7fb84263ea7eef2ce100f24905c444a2c"}]},"branch":"refs/heads/master"},"7535e35771d4089c6179f3661c02ce2405de4a8f":{"kind":"REWORK","_number":2,"created":"2019-09-10 13:57:18.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/2"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:56:43.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7535e35771d4089c6179f3661c02ce2405de4a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7535e35771d4089c6179f3661c02ce2405de4a8f"}]},"branch":"refs/heads/master"},"4a88e1ee7ea1a430d1b4584b32d219dee42f2f57":{"kind":"REWORK","_number":3,"created":"2019-09-10 14:03:14.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/3"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:02:54.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a88e1ee7ea1a430d1b4584b32d219dee42f2f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a88e1ee7ea1a430d1b4584b32d219dee42f2f57"}]},"branch":"refs/heads/master"},"dc33ba6b95e0467a03bd121b2cb17f6fe526f147":{"kind":"REWORK","_number":4,"created":"2019-09-10 14:10:33.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/4"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:10:33.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc33ba6b95e0467a03bd121b2cb17f6fe526f147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc33ba6b95e0467a03bd121b2cb17f6fe526f147"}]},"branch":"refs/heads/master"},"45b2b4e1904d13d2da941f419222bb87de214124":{"kind":"REWORK","_number":5,"created":"2019-09-10 14:34:12.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/5"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:34:05.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45b2b4e1904d13d2da941f419222bb87de214124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45b2b4e1904d13d2da941f419222bb87de214124"}]},"branch":"refs/heads/master"},"8c2015788d3cac23c7e0e2db2d2dba1d7c831a34":{"kind":"REWORK","_number":6,"created":"2019-09-10 14:35:43.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/6"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:35:21.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c2015788d3cac23c7e0e2db2d2dba1d7c831a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c2015788d3cac23c7e0e2db2d2dba1d7c831a34"}]},"branch":"refs/heads/master"},"c1a5c816c7ef4152adf8a9d45b120cdb0f6c107c":{"kind":"REWORK","_number":7,"created":"2019-09-10 14:37:20.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/7"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:37:12.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a5c816c7ef4152adf8a9d45b120cdb0f6c107c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a5c816c7ef4152adf8a9d45b120cdb0f6c107c"}]},"branch":"refs/heads/master"},"d3d29cf96c1c82ea41403ea3ef2efe24ffca42cf":{"kind":"REWORK","_number":8,"created":"2019-09-10 14:43:16.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/8"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:43:06.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3d29cf96c1c82ea41403ea3ef2efe24ffca42cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3d29cf96c1c82ea41403ea3ef2efe24ffca42cf"}]},"branch":"refs/heads/master"},"995ef5c0b35d5c33a629be2ea32692cd9bf881ab":{"kind":"NO_CHANGE","_number":9,"created":"2019-09-10 15:14:24.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/681259/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/681259/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/681259/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/681259/9"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 13:46:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:55:01.000000000","tz":-240},"subject":"End project gating for charm-neutron-api-genericswitch","message":"End project gating for charm-neutron-api-genericswitch\n\nThis patch ends project gating for the purposes of retiring\nthe charm-neutron-api-genericswitch project.\n\nChange-Id: I6118e54a39c6d7ba1f8ea6c697d7fb4297107b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/995ef5c0b35d5c33a629be2ea32692cd9bf881ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/995ef5c0b35d5c33a629be2ea32692cd9bf881ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
