)]}'
{"id":"openstack%2Fproject-config~681270","triplet_id":"openstack%2Fproject-config~master~Ib92bbfa23f424718745d52dc272f2a0e0af34c0f","project":"openstack/project-config","branch":"master","topic":"retire-charm-neutron-api-genericswitch","hashtags":[],"change_id":"Ib92bbfa23f424718745d52dc272f2a0e0af34c0f","subject":"Remove charm-neutron-api-genericswitch from infra","status":"MERGED","created":"2019-09-10 14:15:39.000000000","updated":"2019-10-01 06:04:28.000000000","submitted":"2019-10-01 06:04:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681270-1569909868329-953ecec0","meta_rev_id":"3ffc44ee5a3384de1cd69571e69eac53de38486b","_number":681270,"virtual_id_number":681270,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 06:04:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"date":"2019-09-23 20:01:19.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-09-10 16:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-23 20:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":2,"date":"2019-10-01 05:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":1,"date":"2019-10-01 05:52:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 16:29:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-23 20:30:55.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-01 05:52:55.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-10-01 06:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ccf59e41e3052eb104725d17fbc32860f4ab9dc6","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:15:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f0c980f64e9f618584d0248f30fc88b47d20d41","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:15:56.000000000","message":"Topic changed from end-infra to retire-charm-neutron-api-genericswitch","accounts_in_message":[],"_revision_number":1},{"id":"57caf63d701938a66f0e033fd01d856e45ef317e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:28:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/711042bbfd5549cd809999bc5eb04434 : SUCCESS in 6m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/420061c04d7443eeb64d08a0f209237e : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"cd15215db51b848c39e55fe8105e3a737b847b1f","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 14:58:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3e4e7d7d31f5c30972513e4b5daacb7bd86a019","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 15:01:23.000000000","message":"Patch Set 2: Code-Review-1\n\nREmove the change from zuul.d/projects.yaml again...","accounts_in_message":[],"_revision_number":2},{"id":"68c9162677801e4b9dac56c1bea958f7b901756f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:13:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2bef47c60c75467786160004855250ea : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c2508eb350394a29af34216928a85376 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0f806f3086a57722e240bfe450fef8c90c093e23","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 15:14:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"919935886b6949560c2f808a6ad1d6b9660bb82e","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-10 15:15:48.000000000","message":"Topic changed from remove-charm-neutron-api-genericswitch to retire-charm-neutron-api-genericswitch","accounts_in_message":[],"_revision_number":3},{"id":"4677cdb681b5bf096c023ea88c24a5f43036959c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:32:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d97917b2ab89447d82e5227d838dd69c : SUCCESS in 7m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/96a18aed0fe749ba83024949b9450a74 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3},{"id":"8adc4e3e1442f30c7a5d9e385bd876fa6352642f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 16:29:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b945f0d3c2d79fcaf248f469d7e4d7484518654c","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-23 20:01:19.000000000","message":"Patch Set 3:\n\nThe depends-on patch has now been merged: https://review.opendev.org/#/c/681262/","accounts_in_message":[],"_revision_number":3},{"id":"7e3c71d77546c9c0557d50c4c3977d34a799a286","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 20:30:55.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"d9707bfe15a992b40c58cd49ef214a226a293f7f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-01 05:52:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"45d19ce1356fd3e73dc90ccdd20237409945a7f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 05:53:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f37fd9b1f24016d67cc4628699e9880a1dd4ca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:04:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4ce1411f2425429cb55c6cdb836ac841 : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/83a91c1201994512915c6c622eec4f2e : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"4a7710fddfe701827c613861077b0c4c370cd843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:04:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a469b3d3071f57c5af7edcc131b53b76303288c0","revisions":{"70494354af0e627be3301d908c09afc8bd852bdb":{"kind":"REWORK","_number":1,"created":"2019-09-10 14:15:39.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/70/681270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/681270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/681270/1"}}},"commit":{"parents":[{"commit":"dc33ba6b95e0467a03bd121b2cb17f6fe526f147","subject":"End project gating for charm-neutron-api-genericswitch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc33ba6b95e0467a03bd121b2cb17f6fe526f147"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:13:00.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:14:10.000000000","tz":-240},"subject":"Remove charm-neutron-api-genericswitch from infra","message":"Remove charm-neutron-api-genericswitch from infra\n\nThis patch removes charm-neutron-api-genericswitch from\ninfrastructure systems for the purpose of retiring the project.\n\nDepends-On: https://review.opendev.org/#/c/681262/\nChange-Id: Ib92bbfa23f424718745d52dc272f2a0e0af34c0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70494354af0e627be3301d908c09afc8bd852bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70494354af0e627be3301d908c09afc8bd852bdb"}]},"branch":"refs/heads/master"},"751b941402eac76d15fbce4f9f0f414c343a0335":{"kind":"REWORK","_number":2,"created":"2019-09-10 14:58:23.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/70/681270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/681270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/681270/2"}}},"commit":{"parents":[{"commit":"41db86992f1b3427cad16174b6ab8e90b8b96308","subject":"Merge \"Add allowed-projects to static publish jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41db86992f1b3427cad16174b6ab8e90b8b96308"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:58:10.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:58:10.000000000","tz":-240},"subject":"Remove charm-neutron-api-genericswitch from infra","message":"Remove charm-neutron-api-genericswitch from infra\n\nThis patch removes charm-neutron-api-genericswitch from\ninfrastructure systems for the purpose of retiring the project.\n\nDepends-On: https://review.opendev.org/#/c/681262/\nChange-Id: Ib92bbfa23f424718745d52dc272f2a0e0af34c0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/751b941402eac76d15fbce4f9f0f414c343a0335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/751b941402eac76d15fbce4f9f0f414c343a0335"}]},"branch":"refs/heads/master"},"a469b3d3071f57c5af7edcc131b53b76303288c0":{"kind":"REWORK","_number":3,"created":"2019-09-10 15:14:24.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/70/681270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/681270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/681270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/681270/3"}}},"commit":{"parents":[{"commit":"995ef5c0b35d5c33a629be2ea32692cd9bf881ab","subject":"End project gating for charm-neutron-api-genericswitch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/995ef5c0b35d5c33a629be2ea32692cd9bf881ab"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 14:58:10.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-09-10 15:13:25.000000000","tz":-240},"subject":"Remove charm-neutron-api-genericswitch from infra","message":"Remove charm-neutron-api-genericswitch from infra\n\nThis patch removes charm-neutron-api-genericswitch from\ninfrastructure systems for the purpose of retiring the project.\n\nDepends-On: https://review.opendev.org/#/c/681262/\nChange-Id: Ib92bbfa23f424718745d52dc272f2a0e0af34c0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a469b3d3071f57c5af7edcc131b53b76303288c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a469b3d3071f57c5af7edcc131b53b76303288c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
