)]}'
{"id":"openstack%2Fproject-config~681582","triplet_id":"openstack%2Fproject-config~master~I094066c1ab3b55ae6cfae99aafff125128cd73a4","project":"openstack/project-config","branch":"master","topic":"static-services","hashtags":[],"change_id":"I094066c1ab3b55ae6cfae99aafff125128cd73a4","subject":"Add base promote job for moving static.o.o","status":"MERGED","created":"2019-09-11 19:44:03.000000000","updated":"2019-11-19 19:38:41.000000000","submitted":"2019-11-19 19:38:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681582-1574192321532-99007b45","meta_rev_id":"0996e967c5fceb2af89f10b339ff5c632521a689","_number":681582,"virtual_id_number":681582,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2019-11-19 07:21:09.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 19:38:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-11-19 19:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-15 16:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-11-19 19:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 06:21:49.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-11-15 16:42:46.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-19 19:32:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-11-19 19:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76bbddd685e49cdcca246d05f991b2ddea972a83","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-11 19:44:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5814ad42fbf737a2e9d2eb15a57aa8abfa177ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 19:56:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bf88f9a297e34092841b80ecbcecb2fe : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"69586ea71baa74abd742fb5522481ae9700cb378","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-12 06:21:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b6c9127064ccaf8e21e7e4b2999eeef478b986a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-12 06:33:52.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\nNeeds further design discussion","accounts_in_message":[],"_revision_number":1},{"id":"852e14996e3c24f08c175e9fb4fc6436a50f7b7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-12 06:38:48.000000000","message":"Patch Set 1:\n\nI can change the other jobs so that we publish *in addition* to AFS...\n\nSo, create \"promote-governance-afs\" jobs. Then add these to governance repo so that we use both - and see that everything works fine. Last step would then rename the new job to old name and removal of extra job. Is that what you have in mind?","accounts_in_message":[],"_revision_number":1},{"id":"356a3fee252a66d55617372d6305389e7ac4218d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 09:39:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de8fbaea10f0de5e474ac7e2eca7101e2ea9183b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 09:41:21.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d1a4e8e9a386155f9562fc1f708d3cab6b1b9090","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 09:44:15.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"785a6b1931481e0a81ad35ca854fc1e64740fc5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 09:52:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e01b21509c044dd29f7458167f817183 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":4},{"id":"6a35ead9b19fd9f7dc3721a52dd736767205ec30","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-15 16:42:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"03ce13a00387b3da26fe19672ef3cab5a2b7869f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-18 20:12:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5f3173bc97e3e801ff39d1b7fc2704635617db23","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-19 07:14:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fb062229067a53d3995d40175f4109e948d5176f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-19 07:21:09.000000000","message":"Patch Set 4:\n\nFixed the issue in https://review.opendev.org/694912 - there are more places that needed updating.","accounts_in_message":[],"_revision_number":4},{"id":"c351ccf6c3e107698c0bef22b6f9fa0ace3f1eaa","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-19 19:32:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9a6cb4422fb756ea64510a80d9f970106d94d4de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:32:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e124938b94b52f0f8c6b18dc7407e8865c282c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:38:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f83a89acfbd5ec823caa2cc428996d3e3fd61a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:38:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/525d99f5149043d2a683e186ec9e5e74 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eb2bb351fa41317ada7381c71de69b950dd1dea3","revisions":{"65306c579a27c52db2403dca7a4e086da541bd07":{"kind":"REWORK","_number":1,"created":"2019-09-11 19:44:03.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/82/681582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/681582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/681582/1"}}},"commit":{"parents":[{"commit":"53e935a7b3117c53fd676df4805ae5df3d8f39f8","subject":"Merge \"FN needs to give a little more juice\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53e935a7b3117c53fd676df4805ae5df3d8f39f8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-11 19:24:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-11 19:36:10.000000000","tz":120},"subject":"Add base promote job for moving static.o.o","message":"Add base promote job for moving static.o.o\n\nAdd base job and secret for moving from publishing to\nstatic.o.o to AFS publishing.\n\nChange-Id: I094066c1ab3b55ae6cfae99aafff125128cd73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/65306c579a27c52db2403dca7a4e086da541bd07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/65306c579a27c52db2403dca7a4e086da541bd07"}]},"branch":"refs/heads/master"},"f3eb592165c462d9917e8430e457f12edd95e2f7":{"kind":"REWORK","_number":2,"created":"2019-09-29 09:39:36.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/82/681582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/681582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/681582/2"}}},"commit":{"parents":[{"commit":"53e935a7b3117c53fd676df4805ae5df3d8f39f8","subject":"Merge \"FN needs to give a little more juice\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53e935a7b3117c53fd676df4805ae5df3d8f39f8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-11 19:24:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 09:35:28.000000000","tz":120},"subject":"Add base promote job for moving static.o.o","message":"Add base promote job for moving static.o.o\n\nAdd base job and secret for moving from publishing to\nstatic.o.o to AFS publishing.\n\nStory: 681582\nChange-Id: I094066c1ab3b55ae6cfae99aafff125128cd73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3eb592165c462d9917e8430e457f12edd95e2f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3eb592165c462d9917e8430e457f12edd95e2f7"}]},"branch":"refs/heads/master"},"3d455ce3c645844bb67d49af4f51d81b4221d7c3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-29 09:41:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/82/681582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/681582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/681582/3"}}},"commit":{"parents":[{"commit":"53e935a7b3117c53fd676df4805ae5df3d8f39f8","subject":"Merge \"FN needs to give a little more juice\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53e935a7b3117c53fd676df4805ae5df3d8f39f8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-11 19:24:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 09:40:54.000000000","tz":120},"subject":"Add base promote job for moving static.o.o","message":"Add base promote job for moving static.o.o\n\nAdd base job and secret for moving from publishing to\nstatic.o.o to AFS publishing.\n\nStory: 2006598\nChange-Id: I094066c1ab3b55ae6cfae99aafff125128cd73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d455ce3c645844bb67d49af4f51d81b4221d7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d455ce3c645844bb67d49af4f51d81b4221d7c3"}]},"branch":"refs/heads/master"},"eb2bb351fa41317ada7381c71de69b950dd1dea3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-29 09:44:15.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/82/681582/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/681582/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/681582/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/681582/4"}}},"commit":{"parents":[{"commit":"53e935a7b3117c53fd676df4805ae5df3d8f39f8","subject":"Merge \"FN needs to give a little more juice\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53e935a7b3117c53fd676df4805ae5df3d8f39f8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-11 19:24:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 09:43:39.000000000","tz":120},"subject":"Add base promote job for moving static.o.o","message":"Add base promote job for moving static.o.o\n\nAdd base job and secret for moving from publishing to\nstatic.o.o to AFS publishing.\n\nStory: 2006598\nTask: 36854\nChange-Id: I094066c1ab3b55ae6cfae99aafff125128cd73a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb2bb351fa41317ada7381c71de69b950dd1dea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb2bb351fa41317ada7381c71de69b950dd1dea3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
