)]}'
{"id":"openstack%2Fproject-config~681731","triplet_id":"openstack%2Fproject-config~master~I9d2efdca197689c9b57a07aff43fce5b297b7549","project":"openstack/project-config","branch":"master","topic":"disable-fn","hashtags":[],"change_id":"I9d2efdca197689c9b57a07aff43fce5b297b7549","subject":"Revert \"Disable fortnebula cloud\"","status":"ABANDONED","created":"2019-09-12 12:29:50.000000000","updated":"2019-09-12 15:09:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"18a672b16403ebbc303bbe53d76daf4822c1269c","_number":681731,"virtual_id_number":681731,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"date":"2019-09-12 14:12:24.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-12 14:24:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-09-12 14:52:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-12 15:00:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2019-09-12 13:58:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-12 13:58:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2019-09-12 12:29:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 12:29:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-09-12 13:58:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-12 13:58:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-12 14:24:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 14:52:53.000000000","updated_by":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"reviewer":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"state":"REVIEWER"},{"updated":"2019-09-12 15:00:03.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"ead7b9c5d956447fae9fa644462c60baa6687066","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-12 12:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"846c5d5764a0882ddaf452d860a7841df62854c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 12:45:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6ba001044437478ea9893163f0e31bc3 : SUCCESS in 7m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/53f75a3278cc4567a7a16cd7811d1a67 : POST_FAILURE in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"077a9e1bf38e672463ecc9f7ed4a3251ffed324f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 13:58:35.000000000","message":"Patch Set 1: Code-Review-1\n\nGiven we know there are IPv6 issues here, maybe we should wait until next week (post feature freeze) for this.","accounts_in_message":[],"_revision_number":1},{"id":"11076019f990c1440233d4a2136152700d19ad50","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 13:58:59.000000000","message":"Patch Set 1: Code-Review-1\n\nFortnebula has been quite unstable for weeks and even more lately. Are we sure all connectivity issues have been solved? This week is feature freeze. Many projects are merging feature patches. Maybe not a good time to try-and-check Fortnebula...","accounts_in_message":[],"_revision_number":1},{"id":"7aea17c25978e5707e92eb1764e3b09b8358d0f8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-12 14:12:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5859e7c2cedc7634320f95d6afe256b4dfadb5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:24:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f60381bfaa18427991e857677ca080dd : SUCCESS in 7m 26s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c45602a2a35749b29c25a938e48af4ab : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ca8679db8b19b657c92497aac263b82bfafd29de","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2019-09-12 14:52:53.000000000","message":"Patch Set 1: Code-Review+1\n\nThe FN network issues were due to IPv6 retransmits. That should be solved with the reduction in MTU for the tenant network. All of the test nodes were being NAT to the mirror via IPv4 which resulted in the unreliable connections. \n\nResolution is to lower the MTU for the tenant network - DONE\nAdditionally move the NAT router into Openstack so connections don\u0027t go all the way up and back down the FN Infrastructe - DONE","accounts_in_message":[],"_revision_number":1},{"id":"a137586bdb27d11cbd8edabae5a091bda80a5ae7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-09-12 15:00:03.000000000","message":"Patch Set 1: Code-Review+1\n\nwe would like to use FN to do some testing that can only be done there so if it can be renabled before FF that woudl be better then after form a nova point of view. i am going to use vexhost to test as much as i can with out FN but we wont be able to test both  multi numa pinning without FN in ci.\n\nwe can do single numa pinning and migration so that will hopefully be good enough and il  manually test multi nuam if i have  time","accounts_in_message":[],"_revision_number":1},{"id":"3581a76f4ac78643bc9a707bf17c8563c74b5316","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-12 15:09:41.000000000","message":"Abandoned\n\nLet\u0027S do it stepwise...","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24e83ce3de0752ea54eea9df6222486b00b825ae","revisions":{"24e83ce3de0752ea54eea9df6222486b00b825ae":{"kind":"REWORK","_number":1,"created":"2019-09-12 12:29:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/31/681731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/681731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/681731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/681731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/681731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/681731/1"}}},"commit":{"parents":[{"commit":"82c4b5aaedc6cb4c07b7df93fd6876bc7a60e323","subject":"Disable fortnebula cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82c4b5aaedc6cb4c07b7df93fd6876bc7a60e323"}]}],"author":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-09-12 12:29:50.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-09-12 12:29:50.000000000","tz":0},"subject":"Revert \"Disable fortnebula cloud\"","message":"Revert \"Disable fortnebula cloud\"\n\nThis reverts commit 82c4b5aaedc6cb4c07b7df93fd6876bc7a60e323.\n\nChange-Id: I9d2efdca197689c9b57a07aff43fce5b297b7549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24e83ce3de0752ea54eea9df6222486b00b825ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24e83ce3de0752ea54eea9df6222486b00b825ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
