)]}'
{"id":"openstack%2Fproject-config~682212","triplet_id":"openstack%2Fproject-config~master~I8d7032a6e836ec26f347d16abd600514089a677b","project":"openstack/project-config","branch":"master","topic":"promote-project-config","hashtags":[],"change_id":"I8d7032a6e836ec26f347d16abd600514089a677b","subject":"Switch project-config to promote jobs [3]: Update building","status":"MERGED","created":"2019-09-14 14:57:43.000000000","updated":"2019-09-17 17:42:07.000000000","submitted":"2019-09-17 17:42:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682212-1568742127826-3473af69","meta_rev_id":"50ed35e478c1b12875c8b647f08590999f3e341a","_number":682212,"virtual_id_number":682212,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-17 17:42:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-17 16:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 08:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-17 16:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-09-17 17:26:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 08:04:10.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-09-17 16:59:01.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-17 17:42:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2fa3ab8df19c4ef7136bdc5292485f9e6571ccb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 14:57:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c29b012c001886d38ebbbf292042efceb922900d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 14:59:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6559b48a6a2b35e6972a845ae3e617a06c2548a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 14:59:50.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"8972fcfa626ade59ae26571272578a5e7ce23101","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 15:01:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df677dffcdd93befdd3896593ec4116c06d74924","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 15:03:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d2fd5fd6ebbdbdc61c07fc111bc92699b8c49277","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 15:17:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/a3b4a63ca24f45b89d6553949483914d : SUCCESS in 2m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d69a30d080cf48cda125abf5fd88f203 : SUCCESS in 6m 06s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/befc6fdd6ef9496d9a6dcfe36f69a0b6 : FAILURE in 2m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f7fe08b7e91b4ac9968d2102b22f95a0 : SUCCESS in 3m 29s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d85a953399a74b99a170469cd216fa8c : SUCCESS in 2m 27s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/2bc3e0d6197a4580a76f318720c590cd : SUCCESS in 2m 38s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/bc4a1de6305844bb93a34f1be4c757b8 : SUCCESS in 6m 06s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/7848dcc86d544916af2c7e5a4727f042 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":4},{"id":"961f59f9a989c58618f10eacc21c075e5aeac8e1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 17:03:15.000000000","message":"Patch Set 4:\n\nproject-config-dib job is fixed in https://review.opendev.org/682218","accounts_in_message":[],"_revision_number":4},{"id":"33037b1d09a8cef6957650ec00b86bf09c1c34bb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 04:47:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4f32ff32221a6edcf7c94aa62b79ce49a378db4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 04:54:51.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/d99a64036e45450db7ea13a7236031f1 : SUCCESS in 2m 17s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f09cc5fb30aa4e9c8e67ad0401afd043 : SUCCESS in 5m 09s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/59c1fe3b687b41fba9e182f1790aafaf : FAILURE in 2m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b491af6f4dfd45bc8e1b16f172a4a43e : SUCCESS in 3m 12s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/6a1b7f7752fa43348a52d26625a5bd89 : SUCCESS in 2m 30s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/ec2bd5959c2146d4bf62c52824b422f2 : SUCCESS in 2m 30s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/fb4e8f95b13547ee94f655c6eed9b956 : SUCCESS in 5m 57s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/458bbb2fbe2c4430ad24455abc8ad6d9 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b7ad32e261c64434d5d4010acdb9c42f59afaced","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 06:58:55.000000000","message":"Patch Set 4:\n\nrecheck - release of dib done","accounts_in_message":[],"_revision_number":4},{"id":"baf767c50bcc6397ffffb748a2ce03e0dc22ec59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 07:14:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/02b2cb3f2e264288bf4ebdefa5692602 : SUCCESS in 3m 56s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/02d6c7e3c212477ab4eb1b7a5a4629c6 : SUCCESS in 6m 12s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/63ed95776b8446d88a8f6aef46c86c88 : SUCCESS in 4m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d309798379d040bc81551f8f96a16288 : SUCCESS in 3m 52s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/925044fb906e4d7197536cf5a6ceceff : SUCCESS in 3m 14s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/57f8331ae95e4fef917cb5bc766435a7 : SUCCESS in 3m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4a5cda9910bc47d1a1df1ee50ee3632f : SUCCESS in 7m 10s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e13f3578d4264aa189f839f547b2efa0 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"29b42c0d47431eeeb7a3d3cb69fe2dc423516ad7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 14:48:16.000000000","message":"Patch Set 4:\n\nrecheck - test that everything is fine after project renames","accounts_in_message":[],"_revision_number":4},{"id":"4478048c2778f4852ff99340c148825ab598e713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 15:00:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/e06db43733c146a2a1c03f4e3247f292 : SUCCESS in 3m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/97fcc3ff71b9434e8e67dc71b4b2cf0c : SUCCESS in 8m 42s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/d1c04cf01cbc475780651ba0aca077c4 : SUCCESS in 5m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e478d69b21fd403b83e9348a4d8fcb6f : SUCCESS in 3m 14s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/bd45ca89a7c348d58cba94bfb4023685 : SUCCESS in 3m 20s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/43bf6a8fde324ba38157ce492afe43f3 : SUCCESS in 3m 02s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1b1a3ff2d1344c859e8fe18c37ca036b : FAILURE in 6m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/5b3767659bfa46a4ba35fc21bd3e00ae : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e796a6df0a6efa31c37746bbe9a66027040e3eba","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 15:03:16.000000000","message":"Patch Set 4:\n\nrecheck - previous recheck was too early","accounts_in_message":[],"_revision_number":4},{"id":"dd102df5323c5cb378a73c423d33c74192b27a47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 15:15:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/7c0979ad8d074d89b969173e85c29cf9 : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ebf69037d30a415d92c346df2f868498 : SUCCESS in 6m 22s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/9265c1570dea4859963e8def5590e97d : SUCCESS in 3m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c7004832f25a4ceaafe1ec51ceacf9f5 : SUCCESS in 3m 47s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/ac49c64b95fd4f8f94dee9f7745a1a60 : SUCCESS in 3m 30s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/c41270a033f64cd6acd85164425bf8fe : SUCCESS in 3m 37s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f3d8f89c35e14d7fb2790b7708dbfc5e : SUCCESS in 6m 39s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/180d9684cf044790873d3dfbc5d1f118 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":4},{"id":"8be17ee5aa5e28799c07eb5d7a3f87f0dd320e81","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-17 08:04:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"471fbfb9e5edb320620207b3b71e665c0c5f4ec9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-17 16:59:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d4d46a1a3b3ead8fe1ad72e8c3171eba0bcb0b6a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 17:26:52.000000000","message":"Patch Set 4: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":4},{"id":"ff7d7973db64a9970eab2ff2079e2197321a3252","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:28:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2fa3992646814ef70dec32d3b5de102ac3bef576","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:42:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/4f7f589bf1f34f59b2af68eb602c200a : SUCCESS in 3m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/601d88606b9041af9191cbdb882e5b76 : SUCCESS in 8m 33s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/3595d1c957ae465685b944a8ac375c30 : SUCCESS in 5m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/071f4a4c77de4c55a3203c0ecedd0be0 : SUCCESS in 5m 22s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/f6ce83872c374460b6308a9b96fdb295 : SUCCESS in 4m 00s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/1ce2f387ccf54177846c59d67fc61e02 : SUCCESS in 2m 55s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a8adadd9d1d64aac9f3d37e256117807 : SUCCESS in 6m 02s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/adcad607187749e79050ef2102c7054b : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"bdd6555427314d7ef5ccbde7fdbc3eb8500d2c09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:42:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc","revisions":{"881a97c59f0fdfff9701b4a0fe3fe4a5cf027db3":{"kind":"REWORK","_number":1,"created":"2019-09-14 14:57:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/682212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/682212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/682212/1"}}},"commit":{"parents":[{"commit":"8fdde11bbf77e094cf4c6082e97452da515e545f","subject":"Switch project-config to promote jobs [1]: Disable publish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8fdde11bbf77e094cf4c6082e97452da515e545f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 14:55:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 14:55:43.000000000","tz":120},"subject":"Switch project-config to promote jobs [3]: Update building","message":"Switch project-config to promote jobs [3]: Update building\n\nFor using opendev-tox-docs, jobs need to publish to a \"html\"\nsubdirectory, update the docs-site and specs jobs.\n\nUpdate .gitignore for this.\n\nRemove unused docs environment from tox.ini.\n\nChange-Id: I8d7032a6e836ec26f347d16abd600514089a677b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/881a97c59f0fdfff9701b4a0fe3fe4a5cf027db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/881a97c59f0fdfff9701b4a0fe3fe4a5cf027db3"}]},"branch":"refs/heads/master"},"fc24eb8d895b217796b36928266b4cf33b553996":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-14 14:59:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/682212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/682212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/682212/2"}}},"commit":{"parents":[{"commit":"67ea3720d9c83b6b99cb0a14793c3a2e6ca7cca4","subject":"Switch project-config to promote jobs [1]: Disable publish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67ea3720d9c83b6b99cb0a14793c3a2e6ca7cca4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 14:55:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 14:58:51.000000000","tz":120},"subject":"Switch project-config to promote jobs [3]: Update building","message":"Switch project-config to promote jobs [3]: Update building\n\nFor using opendev-tox-docs, jobs need to publish to a \"html\"\nsubdirectory, update the docs-site and specs jobs.\n\nUpdate .gitignore for this.\n\nRemove unused docs environment from tox.ini.\n\nDepends-On: https://review.opendev.org/682213\nChange-Id: I8d7032a6e836ec26f347d16abd600514089a677b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc24eb8d895b217796b36928266b4cf33b553996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc24eb8d895b217796b36928266b4cf33b553996"}]},"branch":"refs/heads/master"},"96cb7c759129c642e9b4faed368aaf453783f6d8":{"kind":"REWORK","_number":3,"created":"2019-09-14 15:01:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/682212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/682212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/682212/3"}}},"commit":{"parents":[{"commit":"67ea3720d9c83b6b99cb0a14793c3a2e6ca7cca4","subject":"Switch project-config to promote jobs [1]: Disable publish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67ea3720d9c83b6b99cb0a14793c3a2e6ca7cca4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 14:55:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:00:38.000000000","tz":120},"subject":"Switch project-config to promote jobs [3]: Update building","message":"Switch project-config to promote jobs [3]: Update building\n\nFor using opendev-tox-docs, jobs need to publish to a \"html\"\nsubdirectory, update the docs-site and specs jobs.\n\nUpdate .gitignore for this.\n\nUpdate python files to use yaml.safe_load to avoid a warning.\n\nRemove unused docs environment from tox.ini.\n\nDepends-On: https://review.opendev.org/682213\nChange-Id: I8d7032a6e836ec26f347d16abd600514089a677b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96cb7c759129c642e9b4faed368aaf453783f6d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96cb7c759129c642e9b4faed368aaf453783f6d8"}]},"branch":"refs/heads/master"},"800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-14 15:03:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/682212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/682212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/682212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/682212/4"}}},"commit":{"parents":[{"commit":"6cb2ca301f09c954c058a0ce0582d2a25568ee4e","subject":"Switch project-config to promote jobs [1]: Disable publish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6cb2ca301f09c954c058a0ce0582d2a25568ee4e"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 14:55:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:03:01.000000000","tz":120},"subject":"Switch project-config to promote jobs [3]: Update building","message":"Switch project-config to promote jobs [3]: Update building\n\nFor using opendev-tox-docs, jobs need to publish to a \"html\"\nsubdirectory, update the docs-site and specs jobs.\n\nUpdate .gitignore for this.\n\nUpdate python files to use yaml.safe_load to avoid a warning.\n\nRemove unused docs environment from tox.ini.\n\nDepends-On: https://review.opendev.org/682213\nChange-Id: I8d7032a6e836ec26f347d16abd600514089a677b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
