)]}'
{"id":"openstack%2Fproject-config~682214","triplet_id":"openstack%2Fproject-config~master~I3c227f4229a0572a532009f270d44059698f53e8","project":"openstack/project-config","branch":"master","topic":"promote-project-config","hashtags":[],"change_id":"I3c227f4229a0572a532009f270d44059698f53e8","subject":"Switch project-config to promote jobs [4]: Add promote jobs","status":"MERGED","created":"2019-09-14 15:49:22.000000000","updated":"2019-09-17 18:30:42.000000000","submitted":"2019-09-17 18:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682214-1568745042385-d6d561dc","meta_rev_id":"4c164aacb0ab80e96d4177ecd79dd88e7e4940ef","_number":682214,"virtual_id_number":682214,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-17 18:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-17 16:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 08:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-17 16:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-09-17 18:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 08:05:05.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-09-17 16:59:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-17 18:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe7f2662e0b9a5477855ebd039ed6a85d501e557","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 15:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba97e57f29f6ad9b0a726bac2aa0edf6dc811e94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 15:51:34.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3787, column 30","accounts_in_message":[],"_revision_number":1},{"id":"52f0ceda0f6195fd11be5b2022ba527da9347afb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 15:54:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bff38c95abbef4dccfd0b271fc10aa91ff247e16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 15:57:53.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3787, column 30","accounts_in_message":[],"_revision_number":2},{"id":"69c416ce995ebe5a2be1c4b0ad355c44f872c234","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:02:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ace90025612251149ccc71fe99f767230b73ae88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:23:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae453e7c39034958a044f4835691749d : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"a39cd00ff5aa83643d5398a939e584b09aeacf33","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-17 08:05:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"75dccf18cba9187361fac2d25a820d141c4bf175","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-17 16:59:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a4a69e876fe66629d9820d933aa8aea87f928526","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 17:12:16.000000000","message":"Patch Set 3: Workflow-1\n\nLet\u0027s test first...","accounts_in_message":[],"_revision_number":3},{"id":"d33f3e164171c3326868c4936414c7bdd1e1dcc2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 18:09:42.000000000","message":"Patch Set 3: Workflow+1\n\nLooks good, let\u0027s use it...","accounts_in_message":[],"_revision_number":3},{"id":"8237c2fb52dcff8762757559d46fbc2780181a4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:10:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4e962fe07eae0033758b307b21148801071e0a2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9ba86448b351e3d86300783d46f72a255b610111","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:30:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c7841642bf84a89a061000a65dc7599 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9de5bbc2243e548c75dfe2769aad44f15952c017","revisions":{"37a24bbc011456bbff956509beca19d6beb52231":{"kind":"REWORK","_number":1,"created":"2019-09-14 15:49:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/14/682214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/682214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/682214/1"}}},"commit":{"parents":[{"commit":"800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc","subject":"Switch project-config to promote jobs [3]: Update building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:24:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:49:09.000000000","tz":120},"subject":"Switch project-config to promote jobs [4]: Add promote jobs","message":"Switch project-config to promote jobs [4]: Add promote jobs\n\nCreate new promote jobs for infra-index and docs-site.\nAlso, add promote-openstack-specs job that we can use for publishing\nall specs site later.\n\nAdd the same file list (copied from openstack-zuul-jobs) for the two\nproject-config promote jobs.\n\nRemove the now obsolete publish jobs.\n\nRemove now obsolete playbooks.\n\nChange-Id: I3c227f4229a0572a532009f270d44059698f53e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37a24bbc011456bbff956509beca19d6beb52231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37a24bbc011456bbff956509beca19d6beb52231"}]},"branch":"refs/heads/master"},"c56faad8561de872123ea7d5739c64aceb45e407":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-14 15:54:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/14/682214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/682214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/682214/2"}}},"commit":{"parents":[{"commit":"800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc","subject":"Switch project-config to promote jobs [3]: Update building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:24:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:53:30.000000000","tz":120},"subject":"Switch project-config to promote jobs [4]: Add promote jobs","message":"Switch project-config to promote jobs [4]: Add promote jobs\n\nCreate new promote jobs for infra-index and docs-site.\nAlso, add promote-openstack-specs job that we can use for publishing\nall specs site later.\n\nAdd the same file list (copied from openstack-zuul-jobs) for the two\nproject-config promote jobs.\n\nRemove the now obsolete publish jobs.\n\nRemove now obsolete playbooks.\n\nNeeded-By: https://review.opendev.org/682215\nChange-Id: I3c227f4229a0572a532009f270d44059698f53e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c56faad8561de872123ea7d5739c64aceb45e407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c56faad8561de872123ea7d5739c64aceb45e407"}]},"branch":"refs/heads/master"},"9de5bbc2243e548c75dfe2769aad44f15952c017":{"kind":"REWORK","_number":3,"created":"2019-09-14 16:02:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/14/682214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/682214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/682214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/682214/3"}}},"commit":{"parents":[{"commit":"800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc","subject":"Switch project-config to promote jobs [3]: Update building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/800ea0aa056efab44ad4ef1c7bfb6973d06b4ffc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:24:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 16:02:28.000000000","tz":120},"subject":"Switch project-config to promote jobs [4]: Add promote jobs","message":"Switch project-config to promote jobs [4]: Add promote jobs\n\nCreate new promote jobs for infra-index and docs-site.\nAlso, add promote-openstack-specs job that we can use for publishing\nall specs site later.\n\nAdd the same file list (copied from openstack-zuul-jobs) for the two\nproject-config promote jobs.\n\nRemove the now obsolete publish jobs.\n\nRemove now obsolete playbooks.\n\nNeeded-By: https://review.opendev.org/682215\nChange-Id: I3c227f4229a0572a532009f270d44059698f53e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9de5bbc2243e548c75dfe2769aad44f15952c017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9de5bbc2243e548c75dfe2769aad44f15952c017"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
