)]}'
{"id":"openstack%2Fproject-config~682217","triplet_id":"openstack%2Fproject-config~master~Ic28f80d6169116a3efdc49df9fc1650297014dea","project":"openstack/project-config","branch":"master","topic":"promote-project-config","hashtags":[],"change_id":"Ic28f80d6169116a3efdc49df9fc1650297014dea","subject":"Switch project-config to promote jobs [3a]: DNM: Test building","status":"ABANDONED","created":"2019-09-14 15:59:57.000000000","updated":"2019-09-17 18:09:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"410c91f752d2443a320b542fd02c965477a634ec","_number":682217,"virtual_id_number":682217,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-14 16:13:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2019-09-14 16:44:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 16:13:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94b2079a01f9e8c84858f56dac3e3d92845245fa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 15:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"937da9c8da68dcad4d5ec62968743d28f628e372","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:00:38.000000000","message":"Topic changed from t to promote-project-config","accounts_in_message":[],"_revision_number":1},{"id":"47d636ac58115f9c175655b1c898c40fc36485bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/7b6b7b4bfb714d71a6877e2f70e8c39f : SUCCESS in 2m 36s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/67578789f25f4285be39cbb3318a847c : SUCCESS in 5m 58s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/85bf61502d6842109abde6a1cbc06644 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":1},{"id":"352258dbacbe38e7ae8fb47d5f057f639a6578fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:44:45.000000000","message":"Patch Set 1: Workflow-1\n\nneeds recheck once dependency is in.","accounts_in_message":[],"_revision_number":1},{"id":"650b62fb0ad3f50443f7752515b965dd518bbb43","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 17:47:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03b0ad3a830273b05d1761ec9b4dbdfdbf98a9b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:58:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/b531716fec9e4c5d82f66701b9c01e73 : SUCCESS in 2m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/36b6ea26528c401599572e801d20b2d0 : SUCCESS in 6m 21s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/8aa58049272140608323f79efcd86b29 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"410c91f752d2443a320b542fd02c965477a634ec","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 18:09:03.000000000","message":"Abandoned\n\ntesting is fine","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b7958ce48bf0afcc7f522920c8557e2f638ba1d","revisions":{"8b7958ce48bf0afcc7f522920c8557e2f638ba1d":{"kind":"REWORK","_number":1,"created":"2019-09-14 15:59:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/17/682217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/682217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/682217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/682217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/682217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/682217/1"}}},"commit":{"parents":[{"commit":"b1905cd3eb7e01b0318d5cefad05caaaf29e7e90","subject":"Merge \"Update horizon grafana dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1905cd3eb7e01b0318d5cefad05caaaf29e7e90"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:58:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 15:58:13.000000000","tz":120},"subject":"Switch project-config to promote jobs [3a]: DNM: Test building","message":"Switch project-config to promote jobs [3a]: DNM: Test building\n\nDO NOT MERGE THIS, this is just for testing.\n\nSince this repo is a trusted repository, we need to test this changes.\n\nDepends-On: https://review.opendev.org/682212\nChange-Id: Ic28f80d6169116a3efdc49df9fc1650297014dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b7958ce48bf0afcc7f522920c8557e2f638ba1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b7958ce48bf0afcc7f522920c8557e2f638ba1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
