)]}'
{"id":"openstack%2Fproject-config~682218","triplet_id":"openstack%2Fproject-config~master~I9870a57831f2a2d5f5d3641a2290d3bc66d0523d","project":"openstack/project-config","branch":"master","topic":"fix-dib","hashtags":[],"change_id":"I9870a57831f2a2d5f5d3641a2290d3bc66d0523d","subject":"Fix dib job","status":"ABANDONED","created":"2019-09-14 16:08:22.000000000","updated":"2019-09-16 04:48:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"35d050486b34e7edebb53d1ff33f537b2881cac9","_number":682218,"virtual_id_number":682218,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-14 16:09:45.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-14 16:23:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-16 00:26:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 16:23:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-16 00:26:24.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"1f51a8fd749ef0e29a4d0ed1e37965cc6b1a33b5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:08:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"988218a8495a706b236d4a1f20fedc072acbd9a1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:09:45.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/682212 shows the failure.\n\nAlternative: Add flake8 to requirements of dib and make a new release","accounts_in_message":[],"_revision_number":1},{"id":"3e32e32d2831c381c01b1d845900e93e556aae55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/0bbb6485bd4f437797ad07c6550a65ed : SUCCESS in 2m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/35837be41c6e4e6aa1f3020e6b8747e3 : SUCCESS in 6m 03s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/daf4040f8cf64bd8aae4b7260160beef : SUCCESS in 2m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/75d900b9b3e342ef9fc2a638914508c6 : SUCCESS in 3m 16s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/9b2a6c5471cb405f9e89fe27c4ed6daf : SUCCESS in 2m 41s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/c6aa14ed285846a08cbfa917784fcbb3 : SUCCESS in 2m 35s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f708593c28c540ff98ba513b95dcee18 : SUCCESS in 5m 41s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/42d8fd1b85a9470a87d353a47e3fe58e : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d0e69ca34b7afd470f647ff9b72f31744012078a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-16 00:26:24.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks, I believe this was introduced with Iccd9dc6f2508127f17712d16dfce6538ff83807e which moved flake8 to test-requirements.txt.  I missed that dib-lint, which is distributed, calls flake8 directly.\n\nI\u0027ve proposed Ic2eee8f5b64c7020e98e0b1ef43a40f24411891a which re-includes this in the distributed dependencies.  However, I\u0027ve made it an uncapped hacking version, which I believe will alleviate issues from the original change over having a weird capped version of flake8 directly that conflicts system-wide.\n\nSo my preference is to fix this upstream, in case others are using dib-lint, thanks","accounts_in_message":[],"_revision_number":1},{"id":"efac15300d272023f28e4aad5031ecfecc780f97","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 04:48:32.000000000","message":"Abandoned\n\nthe dib change is fine, thanks Ian","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bd24dec7bf932a5b7a09a084b28714a11fc3101","revisions":{"3bd24dec7bf932a5b7a09a084b28714a11fc3101":{"kind":"REWORK","_number":1,"created":"2019-09-14 16:08:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/18/682218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/682218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/682218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/682218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/682218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/682218/1"}}},"commit":{"parents":[{"commit":"b1905cd3eb7e01b0318d5cefad05caaaf29e7e90","subject":"Merge \"Update horizon grafana dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1905cd3eb7e01b0318d5cefad05caaaf29e7e90"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 16:05:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-14 16:08:14.000000000","tz":120},"subject":"Fix dib job","message":"Fix dib job\n\ndiskimage-builder now runs flake8, ensure that it\u0027s installed\nwith installing hacking.\n\nChange-Id: I9870a57831f2a2d5f5d3641a2290d3bc66d0523d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bd24dec7bf932a5b7a09a084b28714a11fc3101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bd24dec7bf932a5b7a09a084b28714a11fc3101"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
