)]}'
{"id":"openstack%2Fproject-config~682544","triplet_id":"openstack%2Fproject-config~master~Ibf7113a73c289112abd2f1a2008666597c2b3737","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ibf7113a73c289112abd2f1a2008666597c2b3737","subject":"Retiring project os-acc.","status":"MERGED","created":"2019-09-17 05:40:07.000000000","updated":"2019-09-20 14:05:27.000000000","submitted":"2019-09-20 14:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682544-1568988327552-953e65f2","meta_rev_id":"931bda450742c87b30d6c22d523be57f0b0fdbc9","_number":682544,"virtual_id_number":682544,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-20 14:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-17 13:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-19 04:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 13:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-20 13:52:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 13:03:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-19 04:40:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-20 13:52:33.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-20 14:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6dbdeaa16741ef9e97143533c02487dd54984ac","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-17 05:40:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bdb3bbac5dabb39c6c13dc57ee6278c42d89eae","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-17 05:51:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4758cd881dcc595f8ded58628851b915bf51ff5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 05:51:59.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"19b512525b6b1e5bad16bad753a316ea5eab5c1d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-17 05:55:24.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"af094b9f188ac77c52a1ffb76048c9e4a8341b8d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 06:00:03.000000000","message":"Patch Set 3:\n\nChange LGTM, will +2 once dependencies are fine and merged","accounts_in_message":[],"_revision_number":3},{"id":"75224c485767a525bc15c29579d7a6a647a0b871","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:05:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b0fc5276dad24716b5a9f177456629df : SUCCESS in 7m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1125d864f7724716a4e97e9a6dd29319 : SUCCESS in 4m 05s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c7afe65f2bdb4f91855d3ea21456dcdf : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a1dd35ccb4bef0fa5114992fe058706b86a6bc9c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-17 13:03:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b38af17e7d7896f6924c65fcb33a2c4370fad50d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-19 04:40:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3c7b62e5f47018e5018b74c5335e12a2c59cff90","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-20 13:52:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f450692cb8be94b7a3fe1f5127631e3730025c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:52:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a7384ee782bf1281a867c09f0521492977253d7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:05:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"931bda450742c87b30d6c22d523be57f0b0fdbc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:05:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e430d432cb04c27b6d8deb4d16b618a : SUCCESS in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e87b840d2708444ea566dabdc61b291d : SUCCESS in 3m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e2dd33caf15e498094cdfc8ed826b68b : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7f42a652ba48fd4d0bfac183b08fb0b0537ee3e","revisions":{"abcfce976a0403d19fc9f79b86d99815b8241cea":{"kind":"REWORK","_number":1,"created":"2019-09-17 05:40:07.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/44/682544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/682544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/682544/1"}}},"commit":{"parents":[{"commit":"5b547bac91d65d52ebb89ddd77a4982dab3437f5","subject":"End gating for os-acc as that project is about to be retired.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b547bac91d65d52ebb89ddd77a4982dab3437f5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 07:33:32.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 07:38:11.000000000","tz":-420},"subject":"Retiring project os-acc.","message":"Retiring project os-acc.\n\nDepends-On: https://review.opendev.org/#/c/682498\nChange-Id: Ibf7113a73c289112abd2f1a2008666597c2b3737\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abcfce976a0403d19fc9f79b86d99815b8241cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abcfce976a0403d19fc9f79b86d99815b8241cea"}]},"branch":"refs/heads/master"},"8d675bb696632b4d1edce65d955f55867900d22e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-17 05:51:00.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/44/682544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/682544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/682544/2"}}},"commit":{"parents":[{"commit":"5b547bac91d65d52ebb89ddd77a4982dab3437f5","subject":"End gating for os-acc as that project is about to be retired.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b547bac91d65d52ebb89ddd77a4982dab3437f5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 07:33:32.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 07:50:00.000000000","tz":-420},"subject":"Retiring project os-acc: Step 4/5.","message":"Retiring project os-acc: Step 4/5.\n\nDepends-On: https://review.opendev.org/#/c/682498\nChange-Id: Ibf7113a73c289112abd2f1a2008666597c2b3737\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d675bb696632b4d1edce65d955f55867900d22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d675bb696632b4d1edce65d955f55867900d22e"}]},"branch":"refs/heads/master"},"a7f42a652ba48fd4d0bfac183b08fb0b0537ee3e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-17 05:55:24.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/44/682544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/682544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/682544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/682544/3"}}},"commit":{"parents":[{"commit":"5b547bac91d65d52ebb89ddd77a4982dab3437f5","subject":"End gating for os-acc as that project is about to be retired.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b547bac91d65d52ebb89ddd77a4982dab3437f5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 07:33:32.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 07:54:18.000000000","tz":-420},"subject":"Retiring project os-acc.","message":"Retiring project os-acc.\n\nStep 4/5 in\nhttps://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nDepends-On: https://review.opendev.org/#/c/682498\nChange-Id: Ibf7113a73c289112abd2f1a2008666597c2b3737\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7f42a652ba48fd4d0bfac183b08fb0b0537ee3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7f42a652ba48fd4d0bfac183b08fb0b0537ee3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
