)]}'
{"id":"openstack%2Fproject-config~682581","triplet_id":"openstack%2Fproject-config~master~I2b7539b78ff31a62a93c9376291122798c177afe","project":"openstack/project-config","branch":"master","topic":"fix-kayobe-acl","hashtags":[],"change_id":"I2b7539b78ff31a62a93c9376291122798c177afe","subject":"Fix release ACL for Kayobe","status":"MERGED","created":"2019-09-17 08:30:13.000000000","updated":"2019-09-18 17:54:30.000000000","submitted":"2019-09-18 17:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682581-1568829270327-805a2d6b","meta_rev_id":"51a95fd34278220d2d16935273e1b1cbc6ca4f52","_number":682581,"virtual_id_number":682581,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-09-17 08:44:06.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-18 17:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-18 17:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-17 08:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-17 08:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-18 17:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 08:31:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-17 08:51:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-18 17:38:51.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-18 17:54:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d92beb1abda664632ed6dc3a7bdf7e3a90d7cfe8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-17 08:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c11254ca3c60d32c10007048119d52ddea7e6c45","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 08:31:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"262f018be9a972af2fd7ee0e82cbf5c412318a5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 08:33:32.000000000","message":"Patch Set 1:\n\nWill it be possible to create releases on stable branches prior to Train if I create deliverables files?","accounts_in_message":[],"_revision_number":1},{"id":"8882399296e6262c147a40da36e362385e0a5aef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/754d30b0d5db41edadb80b464136451e : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d63f681601344fcb8a1be9609fff2792 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"35046b87e0b9b235829d894698471d7206cd122d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-17 08:44:06.000000000","message":"Patch Set 1:\n\nFor those we\u0027ll likely continue to push tags directly (so that we don\u0027t rewrite history as to what was included in previous releases). In order to avoid crazy branch-specific ACLs, Release managers can do that for you if you ask on #openstack-release.","accounts_in_message":[],"_revision_number":1},{"id":"c188c7f3a98009fff091bc48449478960ccd9b38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 08:51:11.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e For those we\u0027ll likely continue to push tags directly (so that we\n \u003e don\u0027t rewrite history as to what was included in previous\n \u003e releases). In order to avoid crazy branch-specific ACLs, Release\n \u003e managers can do that for you if you ask on #openstack-release.\n\nWorks for me. Thanks","accounts_in_message":[],"_revision_number":1},{"id":"46283b1b435a88bb8759b425397a7b3a22df7863","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-18 17:38:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29bda16b0181d77bd246af9022d41c190367730c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:41:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01c917ea17a12eb06c16cdb3d49fc74e44b1866b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:54:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ebb11348a64434fb8e5d13735cc5698 : SUCCESS in 6m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/135a574d328f4d06a4634b4525fbd0a7 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"778631feb38c63d47caf06969379be3e6912204b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:54:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a29dd060553e74ef6eac5cc760e2990135c3b5","revisions":{"d7a29dd060553e74ef6eac5cc760e2990135c3b5":{"kind":"REWORK","_number":1,"created":"2019-09-17 08:30:13.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/81/682581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/682581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/682581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/682581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/682581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/682581/1"}}},"commit":{"parents":[{"commit":"85908f506de206aa0a6f4813e1c9c0ea7f214f07","subject":"Merge \"Fedora 30 : add python-path for Python 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/85908f506de206aa0a6f4813e1c9c0ea7f214f07"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-09-17 08:28:46.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-09-17 08:28:46.000000000","tz":120},"subject":"Fix release ACL for Kayobe","message":"Fix release ACL for Kayobe\n\nKayobe is now an official deliverable from the Kolla project team,\nand as such should go through the openstack/releases repository to\nproduce releases.\n\nFix the ACL so that it no longer allows direct tagging.\n\nChange-Id: I2b7539b78ff31a62a93c9376291122798c177afe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7a29dd060553e74ef6eac5cc760e2990135c3b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7a29dd060553e74ef6eac5cc760e2990135c3b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
