)]}'
{"id":"openstack%2Fproject-config~682711","triplet_id":"openstack%2Fproject-config~master~I2b3b67332093e12a2f177e177715da9d9a9d00f1","project":"openstack/project-config","branch":"master","topic":"test-promote","hashtags":[],"change_id":"I2b3b67332093e12a2f177e177715da9d9a9d00f1","subject":"Update infra and specs indices","status":"MERGED","created":"2019-09-17 18:28:21.000000000","updated":"2019-09-17 20:17:34.000000000","submitted":"2019-09-17 20:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682711-1568751342423-795c03e7","meta_rev_id":"2e218336a51e997e1a82b7b6835ae007f388654c","_number":682711,"virtual_id_number":682711,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-17 20:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-17 18:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 20:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-17 20:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 18:29:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-17 20:05:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-17 20:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69eec900d68dfaf396c35a313aaa3512f47c0803","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 18:28:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f50481425185ceb0e08f8ef9083f2421a99b8ef","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-17 18:29:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e908daf5ef13c046b38a0eaba5566df599c56cce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:45:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/0f1958b8bf5e45d198cbdbf7645177f6 : SUCCESS in 3m 53s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/89cf500f1f2e4fe7a27bde83dc6467da : SUCCESS in 9m 21s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/6ccf4669626648d7bbfd9703a061f498 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"411d53be45f481a2bb1b810d451922048524c433","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-17 20:05:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2f08ceb27d2217dbf8d5d8cd25d93c32294cee9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:05:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62bd666fdd7178bb60b4f340a551c81773677304","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:15:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/e142357ba60d4754ae51e0db26e8120d : SUCCESS in 3m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd2837c7a71d400084893c060d35ba47 : SUCCESS in 6m 42s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/b701e326d1b046a8a1cf7b35ced53013 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"514c346e6a318f2802180f5ca735bc767567d8a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:15:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fdd18921bcf1570d6022ee9838daa784492a674","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:17:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/5770d58769f948ca866149d2f9c6d30e : SUCCESS in 1m 29s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/91f57f95f3944ccbbfb58dd9095b6232 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1450cbdf62019ca00cd515eb999c65eaf0c0708f","revisions":{"1450cbdf62019ca00cd515eb999c65eaf0c0708f":{"kind":"REWORK","_number":1,"created":"2019-09-17 18:28:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/11/682711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/682711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/682711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/682711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/682711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/682711/1"}}},"commit":{"parents":[{"commit":"479bc05d563982f0fd69cf77dcf5164007377668","subject":"Merge \"Switch project-config to promote jobs [3]: Update building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/479bc05d563982f0fd69cf77dcf5164007377668"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-17 18:26:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-17 18:26:27.000000000","tz":120},"subject":"Update infra and specs indices","message":"Update infra and specs indices\n\nAdd missing entries.\n\nThis also tests that publishing works again.\n\nChange-Id: I2b3b67332093e12a2f177e177715da9d9a9d00f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1450cbdf62019ca00cd515eb999c65eaf0c0708f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1450cbdf62019ca00cd515eb999c65eaf0c0708f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
