)]}'
{"id":"openstack%2Fproject-config~682759","triplet_id":"openstack%2Fproject-config~master~I03969da4595f4c66db36cab35ef96a33c6766002","project":"openstack/project-config","branch":"master","topic":"fix-gerrit-config-checks","hashtags":[],"change_id":"I03969da4595f4c66db36cab35ef96a33c6766002","subject":"Fix the zuul main.yaml config checks","status":"MERGED","created":"2019-09-17 22:31:40.000000000","updated":"2019-09-17 23:41:28.000000000","submitted":"2019-09-17 23:41:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682759-1568763688173-14beed78","meta_rev_id":"6b42572f0ebfe13758107ee44cdcd62aecafcb1c","_number":682759,"virtual_id_number":682759,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-09-17 23:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-09-17 22:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 23:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-17 23:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 22:41:10.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-09-17 23:26:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-17 23:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76ede1b99a16de68de3bfd239d371d19f3a9c6e2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-17 22:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74f4797d27c5153c0f2975cbdedd995ae00a4f35","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-17 22:38:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a819832dc9fb0049ab0a63aa657072f41ee55395","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-17 22:40:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64b624f2d81ea99bb92ffb2015d05795ea9a55e7","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-17 22:41:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"03753a0163e57c6e98a95fd54216f9fad5b0d542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:52:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cb20b9b309c848d08b6e86722dc652e8 : SUCCESS in 6m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/880d817483f040b8a37212e4ab61dd1e : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9e5ac0379a68a90f4fe49ab98d95047ba00e166f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-17 23:26:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7044eb1154d71573565358b378fce45712d8ca06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 23:27:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"951d30e4256ba15d41f58d7e1ed5381fb9680c84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 23:41:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/98182805fed24b4fbcca68a7220064a4 : SUCCESS in 6m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e330e5f91b4d4ca4bfffe1dfcc4d4762 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6d08760335fa5960c5d96e73c752193d9c04ef9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 23:41:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a94d0debf21acef5f4d2f628c5d99928182525f","revisions":{"ec78e11ad63acb3f48e9c07b0ae084a7aa808acd":{"kind":"REWORK","_number":1,"created":"2019-09-17 22:31:40.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/59/682759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/682759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/682759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/682759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/682759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/682759/1"}}},"commit":{"parents":[{"commit":"aedf049539ceec954b254999c9feac5e80a1caaf","subject":"Merge \"Update infra and specs indices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aedf049539ceec954b254999c9feac5e80a1caaf"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-09-17 22:30:39.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-09-17 22:30:39.000000000","tz":-420},"subject":"Fix the zuul main.yaml config checks","message":"Fix the zuul main.yaml config checks\n\nIt is possible to have two different types of dict listings for zuul\ntenant project lists. The first (which we handled fine) is a dict per\nproject where we list which items we include from that project. The\nother is a dict with global settings and a list of projects. Handle this\nsecond case.\n\nChange-Id: I03969da4595f4c66db36cab35ef96a33c6766002\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec78e11ad63acb3f48e9c07b0ae084a7aa808acd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec78e11ad63acb3f48e9c07b0ae084a7aa808acd"}]},"branch":"refs/heads/master"},"4a94d0debf21acef5f4d2f628c5d99928182525f":{"kind":"REWORK","_number":2,"created":"2019-09-17 22:40:13.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/59/682759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/682759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/682759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/682759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/682759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/682759/2"}}},"commit":{"parents":[{"commit":"aedf049539ceec954b254999c9feac5e80a1caaf","subject":"Merge \"Update infra and specs indices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aedf049539ceec954b254999c9feac5e80a1caaf"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-09-17 22:30:39.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-09-17 22:40:03.000000000","tz":-420},"subject":"Fix the zuul main.yaml config checks","message":"Fix the zuul main.yaml config checks\n\nIt is possible to have two different types of dict listings for zuul\ntenant project lists. The first (which we handled fine) is a dict per\nproject where we list which items we include from that project. The\nother is a dict with global settings and a list of projects. Handle this\nsecond case.\n\nChange-Id: I03969da4595f4c66db36cab35ef96a33c6766002\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a94d0debf21acef5f4d2f628c5d99928182525f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a94d0debf21acef5f4d2f628c5d99928182525f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
