)]}'
{"id":"openstack%2Fproject-config~682771","triplet_id":"openstack%2Fproject-config~master~I3dfe7ac8c84ddd70b7b10b0ea2c5a30de0b81d94","project":"openstack/project-config","branch":"master","topic":"gerritbot-neutron-stadium","hashtags":[],"change_id":"I3dfe7ac8c84ddd70b7b10b0ea2c5a30de0b81d94","subject":"Stop gerrit notification to neutron-stadium specific channels","status":"MERGED","created":"2019-09-18 01:40:31.000000000","updated":"2019-09-18 14:02:12.000000000","submitted":"2019-09-18 14:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682771-1568815332327-d2cb535d","meta_rev_id":"70004143d2fb0152b5af368c38cb16d8a57bb6f2","_number":682771,"virtual_id_number":682771,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-18 14:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-09-18 13:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 13:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-18 13:46:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 13:04:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-18 13:46:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-18 14:02:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"620d509ab730e62e2a13307cbaf7748c1cb255b6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-18 01:40:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b3f98b2463bd61f408b367c1d02d86c500e7bf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:53:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/387c9b16bbf14afebe1b3e1b7f462387 : SUCCESS in 6m 55s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7a8c0322006848ff9f8ab13cb92983d3 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a59195b3c9be88ddaa89e197804839b9184342aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 04:31:08.000000000","message":"Patch Set 1:\n\nPlease remove channels from hiera/common.yaml in opendev/system-config as well - and if really unneeded here from accessbot","accounts_in_message":[],"_revision_number":1},{"id":"1b3c9545cce2483e9e8f6e4e57bfc57121513906","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 13:04:30.000000000","message":"Patch Set 1: Code-Review+2\n\nNo need to update accessbot, that can stay as is.","accounts_in_message":[],"_revision_number":1},{"id":"3272cee8b64aba4f41f930011564d3f2ecc8756b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-18 13:46:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"763a06d30206a9e609a0d445ad8fb7d1326c06ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:47:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e695733da641ed9e55195fb7ba95e287c3e5e1b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:02:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"febf76649bd81d4c79d66898f573c73c8cb44f99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:02:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3f5ad02b58f34ff0b149152a30c67fe0 : SUCCESS in 7m 43s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a0a2f924e3ba45499b4d62510df02c23 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0113ab3d627446b0a3eb012884dc0e704ac2a9d9","revisions":{"0113ab3d627446b0a3eb012884dc0e704ac2a9d9":{"kind":"REWORK","_number":1,"created":"2019-09-18 01:40:31.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/71/682771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/71/682771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/682771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/682771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/682771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/71/682771/1"}}},"commit":{"parents":[{"commit":"aedf049539ceec954b254999c9feac5e80a1caaf","subject":"Merge \"Update infra and specs indices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aedf049539ceec954b254999c9feac5e80a1caaf"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-18 01:36:02.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-18 01:39:24.000000000","tz":540},"subject":"Stop gerrit notification to neutron-stadium specific channels","message":"Stop gerrit notification to neutron-stadium specific channels\n\nopenstack-fwaas, openstack-vpnaas and networking-sfc IRC channels\nare less active and have a smaller number of subscribers.\nAll notifications are now sent to openstack-neutron channel.\n\nThis is based on the openstack-discuss thread.\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-September/thread.html#9395\n\nChange-Id: I3dfe7ac8c84ddd70b7b10b0ea2c5a30de0b81d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0113ab3d627446b0a3eb012884dc0e704ac2a9d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0113ab3d627446b0a3eb012884dc0e704ac2a9d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
