)]}'
{"id":"openstack%2Fproject-config~682934","triplet_id":"openstack%2Fproject-config~master~Id943aa038c5321074911d4678d25a1f41cd718df","project":"openstack/project-config","branch":"master","topic":"prepare-workspace-git","hashtags":[],"change_id":"Id943aa038c5321074911d4678d25a1f41cd718df","subject":"Symlink /opt/git to /opt/git/opendev.org","status":"MERGED","created":"2019-09-18 15:49:07.000000000","updated":"2019-09-18 16:30:14.000000000","submitted":"2019-09-18 16:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682934-1568824214021-3f030d00","meta_rev_id":"50785fecac205758113ed71c193c880113e01f04","_number":682934,"virtual_id_number":682934,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-18 16:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-18 16:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-18 16:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-09-18 16:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 15:53:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-18 16:10:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-18 16:12:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-18 16:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f37d267253beca0c78e8f481e4d39decfd02a85","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-18 15:49:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c9c4e6a8a46ca72ece7373b7adb10442983731d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-18 15:53:44.000000000","message":"Patch Set 1: Code-Review+2\n\nAccording to https://docs.openstack.org/diskimage-builder/latest/developer/developing_elements.html this is within the chroot, so should work.","accounts_in_message":[],"_revision_number":1},{"id":"165bed05304ffa18fed7047dd56045595c4855bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 15:59:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b66bc0058984474ba64a6e8a6818a35 : SUCCESS in 5m 53s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/d9b5d9dab49447afa84a00f029e1e46b : FAILURE in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"89f3a298acb3089ebddf226d97368aa8fd1527f9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-18 16:03:29.000000000","message":"Patch Set 1:\n\nThis fails CI because of: https://zuul.opendev.org/t/openstack/build/d9b5d9dab49447afa84a00f029e1e46b/log/job-output.txt#509\n\nI believe the fix is to add a header like:\nhttps://opendev.org/openstack/project-config/src/branch/master/nodepool/elements/nodepool-base/install.d/20-iptables#L17-L21","accounts_in_message":[],"_revision_number":1},{"id":"909a794efd9466e6391d272d653dfcf42e5be823","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-18 16:09:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46611942bf01730db853d310a7fc30d132b3276f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-18 16:10:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eec280a94161a8ec4ab3befdedc0258476794ad8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 16:12:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aaa8dbb01c4ebed712db800a9518ee04bca9380f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:22:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f5fa5f1b2136408bb3621d8c813a2819 : SUCCESS in 6m 39s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6b615191b4984e02a7d3f368f79b9058 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1cb8da21ab13f82de302dda0148f0eb3eae76200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:22:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65f1182a743d5a78788ce7f3ed0f24383a46ce5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:30:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3d4d896ba3a5435fa1931779f544ae89 : SUCCESS in 6m 38s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/531bb92c9e9643c29f15c910fc5b0a89 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4f54a7dc918cf290d72dc42fa569417aeb60b3c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:30:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bfed3d564a194ef9706f7731a4a34ded7b67f57e","revisions":{"252b4523d0e85633467b59d0b263be6eaaafda1e":{"kind":"REWORK","_number":1,"created":"2019-09-18 15:49:07.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/34/682934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/682934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/682934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/682934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/682934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/682934/1"}}},"commit":{"parents":[{"commit":"278b7a39e2d572146d30e2e4e6b9322dd815245d","subject":"Merge \"Stop gerrit notification to neutron-stadium specific channels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/278b7a39e2d572146d30e2e4e6b9322dd815245d"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:23.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:23.000000000","tz":-420},"subject":"Symlink /opt/git to /opt/git/opendev.org","message":"Symlink /opt/git to /opt/git/opendev.org\n\nThis will let us move the cached repo location to match what\nprepare-workspace-git expects.\n\nChange-Id: Id943aa038c5321074911d4678d25a1f41cd718df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/252b4523d0e85633467b59d0b263be6eaaafda1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/252b4523d0e85633467b59d0b263be6eaaafda1e"}]},"branch":"refs/heads/master"},"bfed3d564a194ef9706f7731a4a34ded7b67f57e":{"kind":"REWORK","_number":2,"created":"2019-09-18 16:09:32.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/34/682934/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/682934/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/682934/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/682934/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/682934/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/682934/2"}}},"commit":{"parents":[{"commit":"278b7a39e2d572146d30e2e4e6b9322dd815245d","subject":"Merge \"Stop gerrit notification to neutron-stadium specific channels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/278b7a39e2d572146d30e2e4e6b9322dd815245d"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:23.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 16:08:02.000000000","tz":-420},"subject":"Symlink /opt/git to /opt/git/opendev.org","message":"Symlink /opt/git to /opt/git/opendev.org\n\nThis will let us move the cached repo location to match what\nprepare-workspace-git expects.\n\nChange-Id: Id943aa038c5321074911d4678d25a1f41cd718df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfed3d564a194ef9706f7731a4a34ded7b67f57e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfed3d564a194ef9706f7731a4a34ded7b67f57e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
