)]}'
{"id":"openstack%2Fproject-config~682935","triplet_id":"openstack%2Fproject-config~master~I8ae788dfc98f5102aefef6b57f2b82e7c50922ac","project":"openstack/project-config","branch":"master","topic":"prepare-workspace-git","hashtags":[],"change_id":"I8ae788dfc98f5102aefef6b57f2b82e7c50922ac","subject":"Cache repos in /opt/git/opendev.org","status":"MERGED","created":"2019-09-18 15:49:07.000000000","updated":"2019-10-17 19:00:49.000000000","submitted":"2019-10-17 19:00:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682935-1571338849146-f62c7b6a","meta_rev_id":"e3f93ab7bf96aff40bc9065f0b8b7c955bc013a1","_number":682935,"virtual_id_number":682935,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 19:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-09-18 16:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-10-17 17:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 16:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-10-17 17:53:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-10-17 18:54:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 15:55:43.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-18 16:04:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-10-17 18:54:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-17 19:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af520f1c899386c4a5512b3c571a6e3d56766af1","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-18 15:49:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"396703b34cef604998db82b55c55d2e8c060dbae","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-18 15:54:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b5829f976b4b22cd182348290dd1e1203fa634fa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-18 15:55:12.000000000","message":"Patch Set 1:\n\nThis needs to be WIP for now, right?","accounts_in_message":[],"_revision_number":1},{"id":"cbb30aa74742c8befe40b3011f182b617486bdeb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-18 15:55:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e3a9b7a8e63c116d85d8639dd3e3cda8dd2a8a59","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-18 15:56:52.000000000","message":"Patch Set 2:\n\nLGTM now that the depends-on has been added.","accounts_in_message":[],"_revision_number":2},{"id":"96e8b9ff23c76251953a7fa1a1f5490727ae6840","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-18 16:04:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"def03b5c73033f3726a9c6695050619c7971f2fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:05:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c17d3f05f4a4b42839ced0083dc8b0f : SUCCESS in 8m 07s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/147ee5446a3442c1aa0cf57aedce49a2 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"553d9bc87102728978396b9181a5407d39583cb0","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-18 16:09:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ab5efce08029894715fd04a547e38fab7ba820b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:22:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f5aa7f8edd344e148842904d4b18d349 : SUCCESS in 6m 00s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/1d35eeadc77b412ea1a7d1ed42e9bf3d : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"46f22f25e41f7ae17be56f2dde4b63ff348ba493","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-15 20:00:08.000000000","message":"Patch Set 3: Workflow-1\n\nWaiting for openstack release","accounts_in_message":[],"_revision_number":3},{"id":"3f7b9021e5a0d9ad9e3651d20c48d50451cd3fb7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-17 17:31:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"526d3c29f82c93ed867586e29c21499f2d17952c","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-17 17:53:13.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"947f7a4cae3055d55ebfce96510acef426d92653","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-17 18:54:00.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"ff83ff9377d9907db8854b41d3f88acc2ac11310","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-17 18:54:07.000000000","message":"Patch Set 3: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":3},{"id":"d881b8207ddb3f3ab10b7ddfd4c9bc7f97ce4ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:54:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e169606eea409413df69f90d8443d5466d54d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 19:00:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9473b7123ed24b928d25d05852b32ea0 : SUCCESS in 5m 36s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/1a497ad811064840989b6220e2458d1f : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f7ddedaf9813fd66d3bbd32d37afbfd938da971c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 19:00:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"56e6f90b7787ded713c5376f5561328c5719e58f","revisions":{"ced71816486ccc1ed3a96c6ac6fb051deef6e315":{"kind":"REWORK","_number":1,"created":"2019-09-18 15:49:07.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/35/682935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/682935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/682935/1"}}},"commit":{"parents":[{"commit":"252b4523d0e85633467b59d0b263be6eaaafda1e","subject":"Symlink /opt/git to /opt/git/opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/252b4523d0e85633467b59d0b263be6eaaafda1e"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:58.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:58.000000000","tz":-420},"subject":"Cache repos in /opt/git/opendev.org","message":"Cache repos in /opt/git/opendev.org\n\nThis is what prepare-workspace-git expects.  This change should be\nmerged after we have used the symlink (removed in this change) to\nmigrate to that role.\n\nChange-Id: I8ae788dfc98f5102aefef6b57f2b82e7c50922ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ced71816486ccc1ed3a96c6ac6fb051deef6e315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ced71816486ccc1ed3a96c6ac6fb051deef6e315"}]},"branch":"refs/heads/master"},"789131f9be38eec6c44d14c69b022ff5699a1ee8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-18 15:54:33.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/35/682935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/682935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/682935/2"}}},"commit":{"parents":[{"commit":"252b4523d0e85633467b59d0b263be6eaaafda1e","subject":"Symlink /opt/git to /opt/git/opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/252b4523d0e85633467b59d0b263be6eaaafda1e"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:58.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"corvus@inaugust.com","date":"2019-09-18 15:54:33.000000000","tz":0},"subject":"Cache repos in /opt/git/opendev.org","message":"Cache repos in /opt/git/opendev.org\n\nThis is what prepare-workspace-git expects.  This change should be\nmerged after we have used the symlink (removed in this change) to\nmigrate to that role.\n\nChange-Id: I8ae788dfc98f5102aefef6b57f2b82e7c50922ac\nDepends-On: https://review.opendev.org/680696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/789131f9be38eec6c44d14c69b022ff5699a1ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/789131f9be38eec6c44d14c69b022ff5699a1ee8"}]},"branch":"refs/heads/master"},"56e6f90b7787ded713c5376f5561328c5719e58f":{"kind":"REWORK","_number":3,"created":"2019-09-18 16:09:32.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/35/682935/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/682935/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/682935/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/682935/3"}}},"commit":{"parents":[{"commit":"bfed3d564a194ef9706f7731a4a34ded7b67f57e","subject":"Symlink /opt/git to /opt/git/opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfed3d564a194ef9706f7731a4a34ded7b67f57e"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 15:47:58.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-09-18 16:09:09.000000000","tz":-420},"subject":"Cache repos in /opt/git/opendev.org","message":"Cache repos in /opt/git/opendev.org\n\nThis is what prepare-workspace-git expects.  This change should be\nmerged after we have used the symlink (removed in this change) to\nmigrate to that role.\n\nChange-Id: I8ae788dfc98f5102aefef6b57f2b82e7c50922ac\nDepends-On: https://review.opendev.org/680696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56e6f90b7787ded713c5376f5561328c5719e58f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56e6f90b7787ded713c5376f5561328c5719e58f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
