)]}'
{"id":"openstack%2Fproject-config~682954","triplet_id":"openstack%2Fproject-config~master~Ib1523234c28bd397d0408e25905fb7905fad1a12","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ib1523234c28bd397d0408e25905fb7905fad1a12","subject":"Set requireContributorAgreement false for windmill roles","status":"MERGED","created":"2019-09-18 17:44:18.000000000","updated":"2019-09-20 14:05:29.000000000","submitted":"2019-09-20 14:05:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682954-1568988329153-a9f074f5","meta_rev_id":"6f43fb1f41a29dc2e10f03fe1bc33fb948203e70","_number":682954,"virtual_id_number":682954,"owner":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-09-20 14:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-09-18 17:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-20 13:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-20 13:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 17:45:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-20 13:54:41.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-20 14:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25dd84eb4bac32648f24dd2c690fa125017f42df","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-09-18 17:44:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53561b33788c08ead239447e5f357678b3d9928a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 17:45:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a54945da274e266f14c67ef5ed19e0096ce6d03d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:56:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3d5f3e96fd8a421babdb3e397871994e : SUCCESS in 6m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4c4e908996ba43cb9758ea3ee690a456 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"77fa8790d9c3ca4ea17c04ea97eb0281d38149f6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-20 13:54:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d56160455d2fbf3eb2fb5d1602b6103b380ec4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:54:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02e86ac53c2bfb9231f19160ba11d1aed50d2794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:05:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/19e4bc575a25423293a432995a7ac495 : SUCCESS in 8m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f46f7a83460242648bf42dcf50d2fdf5 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6f43fb1f41a29dc2e10f03fe1bc33fb948203e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:05:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cdb57797803e7ee98a40a89e051a09348794056","revisions":{"7cdb57797803e7ee98a40a89e051a09348794056":{"kind":"REWORK","_number":1,"created":"2019-09-18 17:44:18.000000000","uploader":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"ref":"refs/changes/54/682954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/682954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/682954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/682954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/682954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/682954/1"}}},"commit":{"parents":[{"commit":"4e44900bf6eb124dd466f8c142f6d2244fd27373","subject":"Add windmill-ops / ansible-role-elasticsearch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e44900bf6eb124dd466f8c142f6d2244fd27373"}]}],"author":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-09-18 17:43:27.000000000","tz":-240},"committer":{"name":"Paul Belanger","email":"pabelanger@redhat.com","date":"2019-09-18 17:43:27.000000000","tz":-240},"subject":"Set requireContributorAgreement false for windmill roles","message":"Set requireContributorAgreement false for windmill roles\n\nWe don\u0027t need to enforce a CLA here. Remove it to be a little more\nfriendly to new contributors.\n\nChange-Id: Ib1523234c28bd397d0408e25905fb7905fad1a12\nSigned-off-by: Paul Belanger \u003cpabelanger@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7cdb57797803e7ee98a40a89e051a09348794056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7cdb57797803e7ee98a40a89e051a09348794056"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
