)]}'
{"id":"openstack%2Fproject-config~682972","triplet_id":"openstack%2Fproject-config~master~I9d1a9680ce00f9233ce73f9bfc7cb64cf3a5fc54","project":"openstack/project-config","branch":"master","topic":"rm-publish-openstack-tox-docs","hashtags":[],"change_id":"I9d1a9680ce00f9233ce73f9bfc7cb64cf3a5fc54","subject":"Remove obsolete docs publish jobs","status":"MERGED","created":"2019-09-18 19:34:39.000000000","updated":"2019-09-20 05:39:35.000000000","submitted":"2019-09-20 05:39:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682972-1568957975727-0d6172b9","meta_rev_id":"54c93290cfaca1fb4073af001a2d782ee8fe1fc5","_number":682972,"virtual_id_number":682972,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-20 05:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-19 19:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 05:30:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-20 05:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 19:39:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-20 05:30:15.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-09-20 05:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e2672ce343f74b627223bb430e7ca1d8fe06476","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 19:34:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b492aed45d1e8c10f6355defca769069e1a0ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:35:05.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/grenade on branch stable/ocata.  The error was:\n\n  Job publish-openstack-tox-docs not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - neutron-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n          - grenade-heat:\n              voting: false\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n          - tox-docs\n          - ironic-grenade-dsvm:\n              voting: false\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n      gate:\n        jobs:\n          - neutron-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n          - tox-docs\n      post:\n        jobs:\n          - publish-openstack-tox-docs\n      experimental:\n        jobs:\n          - legacy-grenade-dsvm-neutron-forward:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n          - trove-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n\n  in \"openstack/grenade/.zuul.yaml@stable/ocata\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"5758175cf78f8f9b2b9e7967f6098c9274e3f34b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 19:42:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a3b8052a3e3d3d910fc71c8380a06e3ac712fc08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:43:22.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tripleo-validations on branch stable/stein.  The error\nwas:\n\n  Job publish-openstack-tox-docs not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n      - tripleo-multinode-container-minimal\n      - openstack-python-jobs\n      - openstack-python35-jobs\n      - openstack-python36-jobs\n      - tripleo-validations-molecule-jobs\n      - check-requirements\n      - release-notes-jobs-python3\n      check:\n        jobs:\n        - openstack-tox-linters\n        - tripleo-validations-docs\n        - openstack-tox-lower-constraints\n        - tripleo-ci-centos-7-scenario004-standalone:\n            files:\n              - ^roles/ceph.*$\n      gate:\n        jobs:\n        - openstack-tox-linters\n        - tripleo-validations-docs\n        - openstack-tox-lower-constraints\n        - tripleo-ci-centos-7-scenario004-standalone:\n            files:\n              - ^roles/ceph.*$\n      post:\n        jobs:\n        - publish-openstack-tox-docs\n\n  in \"openstack/tripleo-validations/zuul.d/layout.yaml@stable/stein\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"cc6f658cd49d473e85b48e57c51f828aa2161fcb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 19:46:10.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"102dd4d44c0804ff37b676842b0d86917b91bfc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 20:00:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de1ac3312c5c4f4aa56ba99143f5fa24 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c1f670e034e7eeee3e91640a9ecc6fd822edcbdd","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-19 19:39:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3fccfa2ce307b4813f92f7470a86104f8c4c3446","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-20 05:30:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9819b82e322335f083b7f07770c091bae6a8c537","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:30:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b76192acf041f92913939a5248ef9b132ac5fd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:39:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"65983f15a5b67169374ed2cd69136ba987340ea9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:39:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1b65d37af8df4e7ea58f01a606affbf3 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee9e45fc1e23ea45ac842bc8e9f759f1c9973769","revisions":{"067c1d81ce39be4aacb5b8728ef056a1651fb308":{"kind":"REWORK","_number":1,"created":"2019-09-18 19:34:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/72/682972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/682972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/682972/1"}}},"commit":{"parents":[{"commit":"415f773b656c60e5ca1a62866995378e03240f40","subject":"Merge \"Add windmill-ops / ansible-role-elasticsearch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415f773b656c60e5ca1a62866995378e03240f40"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:33:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:33:01.000000000","tz":120},"subject":"Remove obsolete docs publish jobs","message":"Remove obsolete docs publish jobs\n\nThese jobs are not used anymore, their usage has been replaced by\npromote jobs. Remove them.\n\nDepends-On: https://review.opendev.org/682971\nDepends-On: https://review.opendev.org/677245\nChange-Id: I9d1a9680ce00f9233ce73f9bfc7cb64cf3a5fc54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/067c1d81ce39be4aacb5b8728ef056a1651fb308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/067c1d81ce39be4aacb5b8728ef056a1651fb308"}]},"branch":"refs/heads/master"},"88a761a4fe15a8e7e90f3a219c4018101617f97b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-18 19:42:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/72/682972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/682972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/682972/2"}}},"commit":{"parents":[{"commit":"415f773b656c60e5ca1a62866995378e03240f40","subject":"Merge \"Add windmill-ops / ansible-role-elasticsearch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415f773b656c60e5ca1a62866995378e03240f40"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:33:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:41:42.000000000","tz":120},"subject":"Remove obsolete docs publish jobs","message":"Remove obsolete docs publish jobs\n\nThese jobs are not used anymore, their usage has been replaced by\npromote jobs. Remove them.\n\nDepends-On: https://review.opendev.org/682971\nDepends-On: https://review.opendev.org/677245\nDepends-On: https://review.opendev.org/682973\nDepends-On: https://review.opendev.org/682974\nDepends-On: https://review.opendev.org/682975\nDepends-On: https://review.opendev.org/682976\nDepends-On: https://review.opendev.org/682978\n\nChange-Id: I9d1a9680ce00f9233ce73f9bfc7cb64cf3a5fc54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a761a4fe15a8e7e90f3a219c4018101617f97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a761a4fe15a8e7e90f3a219c4018101617f97b"}]},"branch":"refs/heads/master"},"ee9e45fc1e23ea45ac842bc8e9f759f1c9973769":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-18 19:46:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/72/682972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/682972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/682972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/682972/3"}}},"commit":{"parents":[{"commit":"415f773b656c60e5ca1a62866995378e03240f40","subject":"Merge \"Add windmill-ops / ansible-role-elasticsearch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415f773b656c60e5ca1a62866995378e03240f40"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:33:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:45:54.000000000","tz":120},"subject":"Remove obsolete docs publish jobs","message":"Remove obsolete docs publish jobs\n\nThese jobs are not used anymore, their usage has been replaced by\npromote jobs. Remove them.\n\nDepends-On: https://review.opendev.org/682971\nDepends-On: https://review.opendev.org/677245\nDepends-On: https://review.opendev.org/682973\nDepends-On: https://review.opendev.org/682974\nDepends-On: https://review.opendev.org/682975\nDepends-On: https://review.opendev.org/682976\nDepends-On: https://review.opendev.org/682978\nDepends-On: https://review.opendev.org/682982\nChange-Id: I9d1a9680ce00f9233ce73f9bfc7cb64cf3a5fc54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee9e45fc1e23ea45ac842bc8e9f759f1c9973769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee9e45fc1e23ea45ac842bc8e9f759f1c9973769"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
