)]}'
{"id":"openstack%2Fproject-config~683738","triplet_id":"openstack%2Fproject-config~master~Icc0202baf9ce145270454b6f3cd2678cc60189f1","project":"openstack/project-config","branch":"master","topic":"opensuse-grafana","hashtags":[],"change_id":"Icc0202baf9ce145270454b6f3cd2678cc60189f1","subject":"Update Grafana dashboard for openSUSE","status":"MERGED","created":"2019-09-20 22:28:58.000000000","updated":"2019-09-24 06:18:30.000000000","submitted":"2019-09-24 06:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683738-1569305910512-91403195","meta_rev_id":"5a6908508038c7a52ba50fb76073d9dce9cb7cd2","_number":683738,"virtual_id_number":683738,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 06:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-09-23 16:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-24 06:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-24 06:07:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 16:56:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-24 06:07:27.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-09-24 06:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c9e8df23a71d77ca71863716e1b9468c463b143","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-20 22:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebf9d37452a965e880bc315f33a1855dbbd2df03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:37:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/61e5faeb96234e03a09f30884f8c675b : SUCCESS in 6m 05s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/6ba8d959fbf44c52a7e967732d583254 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5bf419c14b92d826effeac60f63effee9b8d6941","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-21 04:18:00.000000000","message":"Patch Set 1:\n\n150 is still in use, isn\u0027t it? So, should we keep it for now?","accounts_in_message":[],"_revision_number":1},{"id":"9ac151795568d406fbcd53c877a0a9c6717e0258","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-22 18:51:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think we should keep 15.0 for now, see http://codesearch.openstack.org/?q\u003dopensuse-150\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nWe still build it.\n\nWant to update 15 to 15.1? See http://grafana.openstack.org/d/NcFtwKriz/nodepool-dib-status?orgId\u003d1","accounts_in_message":[],"_revision_number":1},{"id":"1b629028188b29eaf32bdf28001e0bcd7bb21db8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-23 16:46:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac8f062ccb9e7bf9f6c9b2b72f0d3b4f189d7cb5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-23 16:56:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"569d026d10bd383829c718e2c77bb0802cdf8a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 16:59:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f52bb75cb2464d8b8cfbe01bec96b5c4 : SUCCESS in 7m 48s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/ff14a1b3c02c45aa867c52ef78a51d32 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"8f9e4f0b22836bfe4745e1ed5d6017576ebe4272","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-09-24 06:07:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba754c2333c40b7634721c7a9e6e9b6f5348ba5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 06:07:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"340cdea247a099ffd8ccb3b09e607f2931d87e87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 06:18:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"22a7f0bb3419681404b4f0bcaf676b8de551eae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 06:18:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/46d1d62d248b4f739f895f3cb7d9b12a : SUCCESS in 8m 40s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/94e3018247354a44ae261cbfa36c6b31 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bdafe19c9a10b6da10e7cf438b234c9dede3e12","revisions":{"2f0f6e0de5a0f139f2378866a5fbf081b31a1b39":{"kind":"REWORK","_number":1,"created":"2019-09-20 22:28:58.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/38/683738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/683738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/683738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/683738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/683738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/683738/1"}}},"commit":{"parents":[{"commit":"7a6101a60ccb4e4586884a377cb1091a513d89c8","subject":"Merge \"Add zuul-registry project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a6101a60ccb4e4586884a377cb1091a513d89c8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-20 22:28:17.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-20 22:28:17.000000000","tz":-420},"subject":"Update Grafana dashboard for openSUSE","message":"Update Grafana dashboard for openSUSE\n\nRemove outdated opensuse-150 node and add opensuse-tumbleweed which was\nmissing.\n\nChange-Id: Icc0202baf9ce145270454b6f3cd2678cc60189f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f0f6e0de5a0f139f2378866a5fbf081b31a1b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f0f6e0de5a0f139f2378866a5fbf081b31a1b39"}]},"branch":"refs/heads/master"},"7bdafe19c9a10b6da10e7cf438b234c9dede3e12":{"kind":"REWORK","_number":2,"created":"2019-09-23 16:46:28.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/38/683738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/683738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/683738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/683738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/683738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/683738/2"}}},"commit":{"parents":[{"commit":"7a6101a60ccb4e4586884a377cb1091a513d89c8","subject":"Merge \"Add zuul-registry project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a6101a60ccb4e4586884a377cb1091a513d89c8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-20 22:28:17.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-23 16:45:06.000000000","tz":-420},"subject":"Update Grafana dashboard for openSUSE","message":"Update Grafana dashboard for openSUSE\n\nFix the openSUSE 15.1 image which was incorrectly labeled as just \"15\"\n(which is the nodeset, not the implementation), and add\nopensuse-tumbleweed which was missing.\n\nChange-Id: Icc0202baf9ce145270454b6f3cd2678cc60189f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bdafe19c9a10b6da10e7cf438b234c9dede3e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bdafe19c9a10b6da10e7cf438b234c9dede3e12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
