)]}'
{"id":"openstack%2Fproject-config~684738","triplet_id":"openstack%2Fproject-config~master~I96cf96f281c86f1f2d2ee2cf90d4ec29a3da4dd6","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I96cf96f281c86f1f2d2ee2cf90d4ec29a3da4dd6","subject":"Add ansible-collections-openstack repo","status":"MERGED","created":"2019-09-25 12:46:46.000000000","updated":"2019-09-25 14:43:16.000000000","submitted":"2019-09-25 14:43:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684738-1569422596012-5622d755","meta_rev_id":"175638e0bb4ccc150c6c27047f9840498d14e71f","_number":684738,"virtual_id_number":684738,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 14:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-09-25 13:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-25 14:31:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-25 14:31:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 13:23:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-25 14:31:26.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-25 14:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f04e8b16b344a6af6c2d23e4867c467efce4322","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-09-25 12:46:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d29eef3bd6f7fce8fb9a8e11e4c9ac4ea1b34966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 12:58:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4d7f3e393aa6405fa643cd7aa837230c : SUCCESS in 6m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2f30fcd94f614260b6e7279bbb65c83c : FAILURE in 3m 02s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/96c9304ff7f347da82fcbb979847f7b8 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7bf8282ed7f74fd728c7de44e32006135dc26f7f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-09-25 13:09:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd55985e103bed72735b662bb2faef11bb42e6bd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-25 13:23:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"92c85610c8de87eda6c4c993947e3b5abcf8a0a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 13:27:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/601092805b9b47d490be97a4c932ec36 : SUCCESS in 7m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f63ff6c7fbba488a870c5e56821f8f64 : SUCCESS in 3m 02s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/29cc8a12aae3429d8ee7ed1bd3d8b66c : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ad5455bc15d1c27f5fcfbf1e78ce32806854b5c7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-25 14:31:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97743e8acbc586b54f7b269acdbc1307f10a5156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:31:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bfa3b820d6608997c75b9b16852fcc9adf687314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:43:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a4424b2a2434c36bd75558293611bea : SUCCESS in 6m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/05827d53de4a4953802223b93260fe94 : SUCCESS in 3m 51s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/2100457cbc3f44409dd0a3a749bf9bc8 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"175638e0bb4ccc150c6c27047f9840498d14e71f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:43:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"736ea1f98d756e2a265552de06a81fba15af7014","revisions":{"d1bdc10d1e6b4cd64517d024626691d1071a7043":{"kind":"REWORK","_number":1,"created":"2019-09-25 12:46:46.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/38/684738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/684738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/684738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/684738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/684738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/684738/1"}}},"commit":{"parents":[{"commit":"415f773b656c60e5ca1a62866995378e03240f40","subject":"Merge \"Add windmill-ops / ansible-role-elasticsearch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415f773b656c60e5ca1a62866995378e03240f40"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-09-25 12:44:36.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-09-25 12:46:41.000000000","tz":-240},"subject":"Add ansible-collections-openstack repo","message":"Add ansible-collections-openstack repo\n\nThis will be owned by the openstacksdk team. Ansible is devolving\ncore modules into \"collections\" - which can be managed anywhere.\nThe openstacksdk team has already been managing the openstack\nmodules in ansible\u0027s github, but this will allow us to keep up\nwith things much better.\n\nChange-Id: I96cf96f281c86f1f2d2ee2cf90d4ec29a3da4dd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1bdc10d1e6b4cd64517d024626691d1071a7043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1bdc10d1e6b4cd64517d024626691d1071a7043"}]},"branch":"refs/heads/master"},"736ea1f98d756e2a265552de06a81fba15af7014":{"kind":"REWORK","_number":2,"created":"2019-09-25 13:09:28.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/38/684738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/684738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/684738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/684738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/684738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/684738/2"}}},"commit":{"parents":[{"commit":"415f773b656c60e5ca1a62866995378e03240f40","subject":"Merge \"Add windmill-ops / ansible-role-elasticsearch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/415f773b656c60e5ca1a62866995378e03240f40"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-09-25 12:44:36.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-09-25 13:09:25.000000000","tz":-240},"subject":"Add ansible-collections-openstack repo","message":"Add ansible-collections-openstack repo\n\nThis will be owned by the openstacksdk team. Ansible is devolving\ncore modules into \"collections\" - which can be managed anywhere.\nThe openstacksdk team has already been managing the openstack\nmodules in ansible\u0027s github, but this will allow us to keep up\nwith things much better.\n\nChange-Id: I96cf96f281c86f1f2d2ee2cf90d4ec29a3da4dd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/736ea1f98d756e2a265552de06a81fba15af7014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/736ea1f98d756e2a265552de06a81fba15af7014"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
