)]}'
{"id":"openstack%2Fproject-config~685597","triplet_id":"openstack%2Fproject-config~master~I46d724c2a2aa63902022101686e6972cbd01e76c","project":"openstack/project-config","branch":"master","topic":"cleanup","hashtags":[],"change_id":"I46d724c2a2aa63902022101686e6972cbd01e76c","subject":"Cleanup publish-openstack-releasenotes jobs","status":"MERGED","created":"2019-09-29 15:13:11.000000000","updated":"2019-11-13 20:08:03.000000000","submitted":"2019-11-13 20:08:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685597-1573675683447-a4121484","meta_rev_id":"1b7d6d858b73b7dbe357243a6c6b1dc844cb7a87","_number":685597,"virtual_id_number":685597,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 20:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-11-13 19:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-13 19:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-11-13 19:55:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2019-11-13 19:52:47.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 19:54:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-11-13 19:55:15.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-11-13 20:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"477454b8a34e359752e2423e766bfe450bf8008e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 15:13:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccd16f2c81dc12464f0e7988ccf701865a626649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 15:13:31.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch stable/pike.  The error\nwas:\n\n  Job publish-openstack-releasenotes not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python35-jobs-neutron\n        - publish-openstack-sphinx-docs\n        - release-notes-jobs\n        - check-requirements\n        - periodic-stable-jobs-neutron\n      check:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - build-openstack-releasenotes\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-networking-bgpvpn-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-networking-bagpipe-dsvm-fullstack:\n              voting: false\n      gate:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - build-openstack-releasenotes\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-networking-bgpvpn-bagpipe:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - publish-openstack-releasenotes\n      periodic-stable:\n        jobs:\n          - build-openstack-sphinx-docs:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@stable/pike\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1d645716f53d176d7489c106c06c8cb53383c136","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 15:34:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1b6c6b5e87164345df8cd473f6e6c4c543da90d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 15:52:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/27656dc033634b2188832c3c6a38e386 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"aae290d2f4492afbfea5aac8cd77e0736d352c87","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-01 06:14:48.000000000","message":"Patch Set 2: Workflow-1\n\ndependencies not merged yet","accounts_in_message":[],"_revision_number":2},{"id":"d2d023f99d0583544b205ad4a75122ee39a3a81c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-13 19:52:47.000000000","message":"Patch Set 2: -Workflow\n\ndependencies are merged","accounts_in_message":[],"_revision_number":2},{"id":"0b06acbaa2b8c890ff65123e11040e174517c6f8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-11-13 19:54:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ffdac7e3f813aebe9ec7091a3d26506a884ed06f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-13 19:55:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"78d72e8c97b741cfa13814c723ad5fa512b2d106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 19:55:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"208fefec1dde4d815bad742b37fc78733ad31cad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 20:08:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/852a3a1689af442a872a5d9d89c7ba95 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6c4efb3bd21273e0a8d240570def023865cdf73c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 20:08:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a853c2cbb48d79543733a73bc937156974b3bea","revisions":{"bff03a921b3bd5d2704ca23bb2e130df348add7a":{"kind":"REWORK","_number":1,"created":"2019-09-29 15:13:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/685597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/685597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/685597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/685597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/685597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/685597/1"}}},"commit":{"parents":[{"commit":"eba70a6f6e9054f1ce286c1d51f8c2591ff298ae","subject":"Merge \"Add ansible-collections-openstack repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eba70a6f6e9054f1ce286c1d51f8c2591ff298ae"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 14:57:09.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:12:40.000000000","tz":120},"subject":"Cleanup publish-openstack-releasenotes jobs","message":"Cleanup publish-openstack-releasenotes jobs\n\nRemove publish-openstack-releasenotes job, it\u0027s not used.\n\nNow we don\u0027t need publish-openstack-releasenotes-base anymore, so merge\npublish-openstack-releasenotes-python3 and it\u0027s parent.\n\nChange-Id: I46d724c2a2aa63902022101686e6972cbd01e76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bff03a921b3bd5d2704ca23bb2e130df348add7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bff03a921b3bd5d2704ca23bb2e130df348add7a"}]},"branch":"refs/heads/master"},"7a853c2cbb48d79543733a73bc937156974b3bea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-29 15:34:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/685597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/685597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/685597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/685597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/685597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/685597/2"}}},"commit":{"parents":[{"commit":"eba70a6f6e9054f1ce286c1d51f8c2591ff298ae","subject":"Merge \"Add ansible-collections-openstack repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eba70a6f6e9054f1ce286c1d51f8c2591ff298ae"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 14:57:09.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:25:38.000000000","tz":120},"subject":"Cleanup publish-openstack-releasenotes jobs","message":"Cleanup publish-openstack-releasenotes jobs\n\nRemove publish-openstack-releasenotes job, it\u0027s not used.\n\nNow we don\u0027t need publish-openstack-releasenotes-base anymore, so merge\npublish-openstack-releasenotes-python3 and it\u0027s parent.\n\nDepends-On: https://review.opendev.org/685598\nDepends-On: https://review.opendev.org/685599\nDepends-On: https://review.opendev.org/685600\nChange-Id: I46d724c2a2aa63902022101686e6972cbd01e76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a853c2cbb48d79543733a73bc937156974b3bea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a853c2cbb48d79543733a73bc937156974b3bea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
