)]}'
{"id":"openstack%2Fproject-config~685616","triplet_id":"openstack%2Fproject-config~master~I722d8848b1a1877d758e30c14537f553b93da168","project":"openstack/project-config","branch":"master","topic":"windmill-docs","hashtags":[],"change_id":"I722d8848b1a1877d758e30c14537f553b93da168","subject":"Bye, bye readthedocs-failing","status":"MERGED","created":"2019-09-29 16:04:01.000000000","updated":"2019-10-02 17:02:47.000000000","submitted":"2019-10-02 17:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685616-1570035767397-fdea6b69","meta_rev_id":"cc48b32dd5b84ca31675b24963e1c31b059ef419","_number":685616,"virtual_id_number":685616,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 17:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"all":[{"value":2,"date":"2019-10-02 13:38:12.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-10-02 16:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-02 16:48:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 13:38:12.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-10-02 16:48:18.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-02 17:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb0297d3b1bfbc04974f0eb95c252052a7e5fd55","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 16:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94e086e2747249d1ed86e10f1924d5817e54f752","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 16:12:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"deee256fd39c584c7e15392f365ed94225cfeba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 16:16:38.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f5c47c2628efe0cd25ef5a0c10bade7319c17a47","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 16:19:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6978fbcd0d73e93b990bbb2486c1e25f93f3b7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 16:21:59.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"031c4761f2979e6c132a106d812393d6a8762b1d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 16:22:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"77a4b5e25d3fdd212dcdc75dcbbbb64c798a9f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 16:34:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/50d04e0c872242a4960636cd554c8388 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a23f04c32663e22bc756c806eed638175ec9a7f9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-02 13:37:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"783e54d86a41e0b6130c9d8a5541940d67a5739a","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-10-02 13:38:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"320acc8ae50e1f0d8da1cccdd8c902401e2b6198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 13:50:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8565ae94ebe14884a82055c68338852f : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"53a902a4e3e7f35b2a77a6e137ccdb54b2f1d9ca","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-02 16:48:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"000418d67bc432cdd2e74da15e3fde58804eb940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:48:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"09067c435d0b012f470685458ce9add4cca3e453","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:02:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/36405d06772f4c88aeee47d2afbf77be : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c6be5c384eb4040ae5f96dcf9036c1cb76e14913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:02:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b4d4c7f300d1aa86577cbec0f706ee9301aad6e","revisions":{"7c947fd4781526886b2a34422e1a435d9be8bede":{"kind":"REWORK","_number":1,"created":"2019-09-29 16:04:01.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/685616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/685616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/685616/1"}}},"commit":{"parents":[{"commit":"9dee7f43b438882791be5aeab4ecdd32e9bd5dfc","subject":"Remove windmill docs jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9dee7f43b438882791be5aeab4ecdd32e9bd5dfc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:58:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:58:47.000000000","tz":120},"subject":"Bye, bye readthedocs-failing","message":"Bye, bye readthedocs-failing\n\nWe introduced the readthedocs-failing template over a year ago so\nthat repos get updated (see [1]).\nThis experiment succeeded for some repos, for others it failed.\n\nTelling teams for over a year about this problem is IMHO enough, so\nlet\u0027s stop this.\n\nThis change replaces the readthedocs-failing template with a template\nthat just builds the docs but does not publish so that we can remove the\nreadthedocs-failing template.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-August/132836.html\n\nChange-Id: I722d8848b1a1877d758e30c14537f553b93da168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c947fd4781526886b2a34422e1a435d9be8bede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c947fd4781526886b2a34422e1a435d9be8bede"}]},"branch":"refs/heads/master"},"b09705350f815493fa00468d4c50acfd2970ed47":{"kind":"REWORK","_number":2,"created":"2019-09-29 16:12:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/685616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/685616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/685616/2"}}},"commit":{"parents":[{"commit":"9dee7f43b438882791be5aeab4ecdd32e9bd5dfc","subject":"Remove windmill docs jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9dee7f43b438882791be5aeab4ecdd32e9bd5dfc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:58:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 16:12:01.000000000","tz":120},"subject":"Bye, bye readthedocs-failing","message":"Bye, bye readthedocs-failing\n\nWe introduced the readthedocs-failing template over a year ago so\nthat repos get updated (see [1]).\nThis experiment succeeded for some repos, for others it failed.\n\nTelling teams for over a year about this problem is IMHO enough, so\nlet\u0027s stop this.\n\nThis change replaces the readthedocs-failing template with a template\nthat just builds the docs but does not publish so that we can remove the\nreadthedocs-failing template.\n\nFor yaql repository, it remove the job, instead change\nhttps://review.opendev.org/685619 is proposed to import the template.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-August/132836.html\n\nChange-Id: I722d8848b1a1877d758e30c14537f553b93da168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b09705350f815493fa00468d4c50acfd2970ed47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b09705350f815493fa00468d4c50acfd2970ed47"}]},"branch":"refs/heads/master"},"5b4d4c7f300d1aa86577cbec0f706ee9301aad6e":{"kind":"NO_CHANGE","_number":3,"created":"2019-10-02 13:37:33.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/685616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/685616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/685616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/685616/3"}}},"commit":{"parents":[{"commit":"f66f99506cbdae40f31698c259e7ee70dd9e2d2f","subject":"Remove windmill docs jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f66f99506cbdae40f31698c259e7ee70dd9e2d2f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:58:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-02 13:37:24.000000000","tz":120},"subject":"Bye, bye readthedocs-failing","message":"Bye, bye readthedocs-failing\n\nWe introduced the readthedocs-failing template over a year ago so\nthat repos get updated (see [1]).\nThis experiment succeeded for some repos, for others it failed.\n\nTelling teams for over a year about this problem is IMHO enough, so\nlet\u0027s stop this.\n\nThis change replaces the readthedocs-failing template with a template\nthat just builds the docs but does not publish so that we can remove the\nreadthedocs-failing template.\n\nFor yaql repository, it remove the job, instead change\nhttps://review.opendev.org/685619 is proposed to import the template.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-August/132836.html\n\nChange-Id: I722d8848b1a1877d758e30c14537f553b93da168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b4d4c7f300d1aa86577cbec0f706ee9301aad6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b4d4c7f300d1aa86577cbec0f706ee9301aad6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
