)]}'
{"id":"openstack%2Fproject-config~686251","triplet_id":"openstack%2Fproject-config~master~I0a9f36b3697f8f191dd19601e15c4b6689da930a","project":"openstack/project-config","branch":"master","topic":"charms-train-placement","hashtags":[],"change_id":"I0a9f36b3697f8f191dd19601e15c4b6689da930a","subject":"New charm and interface for placement split","status":"MERGED","created":"2019-10-02 19:07:32.000000000","updated":"2019-10-09 14:38:01.000000000","submitted":"2019-10-09 14:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686251-1570631881722-aecd3d01","meta_rev_id":"094fa4c1509e9f977a16f0e0832d90cdf3880af6","_number":686251,"virtual_id_number":686251,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 14:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-07 12:26:46.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-10-09 14:27:06.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-10-08 17:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-04 13:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2019-10-08 14:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-10-08 14:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":1,"date":"2019-10-08 14:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2019-10-08 15:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-10-09 14:27:06.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 13:55:18.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2019-10-08 14:54:47.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-08 14:55:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-10-08 14:56:35.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-10-08 15:24:57.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-10-08 17:31:44.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-09 14:27:06.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-10-09 14:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7bd901af76b43f15f46346cb99e14191f519fd","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-10-02 19:07:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ae558de481267e1b190236a3b9513eb62731e55","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-10-02 19:09:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3b0cf6b88b2e02787e2ab873ffbf4ceef689e86c","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-10-02 19:09:57.000000000","message":"Topic set to charms-train","accounts_in_message":[],"_revision_number":2},{"id":"32d204e4f508687e5423326ff16e1512bd0c4ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:23:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/10df7941ace84b81a9ae0e5df5ae869a : SUCCESS in 10m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ad2e3f8074224a0f9f5ca58071948e64 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5d1cdccd41e15c6f317fe25db776ead1c78491cd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-03 05:19:30.000000000","message":"Patch Set 2:\n\nLGTM, needs PTL+1","accounts_in_message":[],"_revision_number":2},{"id":"8da8b3924c771af85ea176cf643f5f5fb2adb689","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-10-04 12:49:50.000000000","message":"Topic changed from charms-train to charms-train-placement","accounts_in_message":[],"_revision_number":2},{"id":"4a1155af2c7945bd7f89ced0eaf8ebc8f8d1a819","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-10-04 13:55:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ec3e8f59dcf19fd117ea7411b4dd4f56a61bcf2a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 14:15:10.000000000","message":"Patch Set 2:\n\nNote that you will start with an empty repository. Your very first\npatch needs to setup the repository and add all infrastructure so that\nthe jobs pass (this can be done with a cookiecutter project).\n\nAlso, you can import an existing repository including it\u0027s history if\nyou have one. Once the repository is created, importing is not\npossible anymore, changes will have to go through the normal review\nprocess.\n\nIf you have a lot of history to import, please use the upstream field\ninstead of creating a repository and then pushing the patches one at a\ntime. Pushing a large number of related patches all at one time causes\nthe CI infrastructure to slow down, which impacts work on all of the\nother projects using it. After repo creation the infra team will not\npush any content to it; this is your only chance to import directly.\n\nPlease confirm that starting with an empty repo is fine.","accounts_in_message":[],"_revision_number":2},{"id":"712406019613560e12ca39cb05a7eba46ceacd66","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-10-07 12:26:46.000000000","message":"Patch Set 2:\n\nHi Andreas, starting with an empty repo is fine. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"659223fd34c4d6559fb725c5b1a85d9482de65fe","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-08 14:54:47.000000000","message":"Patch Set 2: Code-Review+1\n\nFine for me. We need this to go forward on that topic in governance.","accounts_in_message":[],"_revision_number":2},{"id":"c406056ec0e87f935d7be9bf457317eaf06c2384","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-08 14:55:39.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm.","accounts_in_message":[],"_revision_number":2},{"id":"36260cb374a964516828c8bb2e380ef64387fe62","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-10-08 14:56:35.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":2},{"id":"017f78b2d0990c8b5686490d2de7045484eddee1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-08 15:16:56.000000000","message":"Patch Set 2:\n\nStill waiting for PTL...","accounts_in_message":[],"_revision_number":2},{"id":"55a1b36773edc641c85f2d766f0e22b9dc814c0c","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-10-08 15:24:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"49f19e760dda2f5d7ecfec1a83851f4cfa3ce6da","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-10-08 16:22:33.000000000","message":"Patch Set 2:\n\n(PTL approval)","accounts_in_message":[],"_revision_number":2},{"id":"d28cea47f9470c90b2f4577da56e2156a35dc393","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-08 17:31:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebf929a357c71c4907efeee7fd82049407499f80","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-09 14:27:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"87b24726bf604dee543e779c98589d6f7e6ddb21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:27:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9be64d4073b97e452e742177ec944a9efa3f6dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:38:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1ba0b785967f4f9fbbfe741b5c3861f1 : SUCCESS in 8m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fda1e3b89845458d92e630a53f714387 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"094fa4c1509e9f977a16f0e0832d90cdf3880af6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:38:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f47f7e477c8dd623ee90ce533a5ad26b95695199","revisions":{"26c052af4ec57974ab3f092a8c08cbc33e4ef82c":{"kind":"REWORK","_number":1,"created":"2019-10-02 19:07:32.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/51/686251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/686251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/686251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/686251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/686251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/686251/1"}}},"commit":{"parents":[{"commit":"5cdd7ac75926bb642dee6691d150d3f45ef471a5","subject":"Merge \"Bye, bye readthedocs-failing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5cdd7ac75926bb642dee6691d150d3f45ef471a5"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-10-02 18:39:34.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-10-02 19:07:31.000000000","tz":0},"subject":"New charm and interface for placement split","message":"New charm and interface for placement split\n\nAdd placement charm and interface that allow for placement API\nto be split from nova-cloud-controller charm.\n\nChange-Id: I0a9f36b3697f8f191dd19601e15c4b6689da930a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/26c052af4ec57974ab3f092a8c08cbc33e4ef82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/26c052af4ec57974ab3f092a8c08cbc33e4ef82c"}]},"branch":"refs/heads/master"},"f47f7e477c8dd623ee90ce533a5ad26b95695199":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-02 19:09:20.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/51/686251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/686251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/686251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/686251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/686251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/686251/2"}}},"commit":{"parents":[{"commit":"5cdd7ac75926bb642dee6691d150d3f45ef471a5","subject":"Merge \"Bye, bye readthedocs-failing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5cdd7ac75926bb642dee6691d150d3f45ef471a5"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-10-02 18:39:34.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-10-02 19:09:10.000000000","tz":0},"subject":"New charm and interface for placement split","message":"New charm and interface for placement split\n\nAdd placement charm and interface that allow for placement API\nto be split from nova-cloud-controller charm.\n\nNeeded-By: https://review.opendev.org/686253\nChange-Id: I0a9f36b3697f8f191dd19601e15c4b6689da930a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f47f7e477c8dd623ee90ce533a5ad26b95695199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f47f7e477c8dd623ee90ce533a5ad26b95695199"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
