)]}'
{"id":"openstack%2Fproject-config~686392","triplet_id":"openstack%2Fproject-config~master~I7922ecef6c4b541ff909d6c3030457b797003685","project":"openstack/project-config","branch":"master","topic":"openapi-sphinx","hashtags":[],"change_id":"I7922ecef6c4b541ff909d6c3030457b797003685","subject":"Add sphinx-contrib/openapi to untrusted github projects","status":"MERGED","created":"2019-10-03 12:45:36.000000000","updated":"2019-10-03 15:18:02.000000000","submitted":"2019-10-03 15:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686392-1570115882702-b42f53c0","meta_rev_id":"3045a0ad6c05f7287c488075a1846a87c3983118","_number":686392,"virtual_id_number":686392,"owner":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-10-03 15:08:25.000000000","_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 15:18:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-10-03 15:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-10-03 12:56:19.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":2,"date":"2019-10-03 15:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":1,"date":"2019-10-03 15:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-03 12:46:16.000000000","_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 12:46:16.000000000","updated_by":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"reviewer":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"state":"REVIEWER"},{"updated":"2019-10-03 12:56:19.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-10-03 15:01:14.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-10-03 15:09:02.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-03 15:18:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ea57dc14cc3668348f7d800246dbc7a82f5c3e3","author":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"date":"2019-10-03 12:45:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cde801e5e97263dedea092f2c7734ea7d7a18a23","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-03 12:56:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97015e8881b51c8fedb8d96a885b975e267fbf5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 12:58:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d07d284b4da425999b53a2a80f64f02 : SUCCESS in 11m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0253a489e90947f78985004ddd00dc20 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7347d930a9af88e1e63399f8dd2b2c94eb819839","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-03 14:58:55.000000000","message":"Patch Set 1:\n\nIs adding these to Zuul tenant really correct?","accounts_in_message":[],"_revision_number":1},{"id":"cb632fbd5dcb1014740caa68065503f84bc96eab","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-03 15:01:07.000000000","message":"Patch Set 1:\n\nZuul does use openapi-sphinx -- I believe Matthieu wants to test enhancements to that.","accounts_in_message":[],"_revision_number":1},{"id":"a43a00ad549d8e69bdaa1d7663177a61480206ec","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-03 15:01:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e808cc125d8f4d741b01495c324de02d32b29ce","author":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"date":"2019-10-03 15:08:25.000000000","message":"Patch Set 1:\n\n\u003e Zuul does use openapi-sphinx -- I believe Matthieu wants to test\n \u003e enhancements to that.\n\nExactly - I have some PRs open on that project to support authentication schemes, which will be used when documenting the admin REST API.","accounts_in_message":[],"_revision_number":1},{"id":"d90741cde41837c209906135796999c5b1d62856","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-03 15:09:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"363c2189ed2eeb6ceadffb98381da29cdafe49ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:09:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84c3dd0d83b3e813eda584fd5c8b9cebda95f9c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:18:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a25ff417bf6e27f201d4b7ac29efb43cd1c555f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:18:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d074e9342aef4165ba09d89a62bbdff6 : SUCCESS in 5m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/df435066a3834772a5787be5bb7bab0e : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce7583352efb5f1a64c6effa7a2a0c442140a6c1","revisions":{"ce7583352efb5f1a64c6effa7a2a0c442140a6c1":{"kind":"REWORK","_number":1,"created":"2019-10-03 12:45:36.000000000","uploader":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"ref":"refs/changes/92/686392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/686392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/686392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/686392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/686392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/686392/1"}}},"commit":{"parents":[{"commit":"f68db8d583e6d8128fbc81b14b6808e525d55eb1","subject":"Add openstacksdk /ara to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f68db8d583e6d8128fbc81b14b6808e525d55eb1"}]}],"author":{"name":"Matthieu Huin","email":"mhuin@redhat.com","date":"2019-10-03 12:44:26.000000000","tz":120},"committer":{"name":"Matthieu Huin","email":"mhuin@redhat.com","date":"2019-10-03 12:44:26.000000000","tz":120},"subject":"Add sphinx-contrib/openapi to untrusted github projects","message":"Add sphinx-contrib/openapi to untrusted github projects\n\nthis will allow testing openAPI docs rendering with Pull Requests\non sphinx-contrib/openapi.\n\nChange-Id: I7922ecef6c4b541ff909d6c3030457b797003685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce7583352efb5f1a64c6effa7a2a0c442140a6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce7583352efb5f1a64c6effa7a2a0c442140a6c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
