)]}'
{"id":"openstack%2Fproject-config~686524","triplet_id":"openstack%2Fproject-config~master~I3414fb9e503f94ff744b560eff9ec0f4afdbb50e","project":"openstack/project-config","branch":"master","topic":"centos8","hashtags":[],"change_id":"I3414fb9e503f94ff744b560eff9ec0f4afdbb50e","subject":"nodepool/elements : use abstracted commands","status":"MERGED","created":"2019-10-04 06:11:49.000000000","updated":"2019-10-10 15:33:09.000000000","submitted":"2019-10-10 15:33:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686524-1570721589468-70f2cce9","meta_rev_id":"908eec83c845978c3b47ef1bc7d6b119690361aa","_number":686524,"virtual_id_number":686524,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:33:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-10-10 04:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-10 06:19:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-10 15:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-10 15:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"date":"2019-10-09 20:44:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 04:42:12.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-10 06:19:33.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-10-10 15:21:04.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-10 15:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd45c34668d5733d9be1368a1110910cbc9e0640","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-04 06:11:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbf957b011f0c1be8e0948e19adfbc72d77d2c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 06:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8f394696526a467294909c433a958133 : SUCCESS in 6m 13s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/bba629a467b941ce8ffdf7399718c6f9 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0b70ec2943227be22cb52205553012163654a90a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-04 06:46:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0549c02ec69bf7ff6a6fb8b106f78c7fd2781767","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-09 03:07:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6204805fc755e613f73c9ea845017ca93cea4bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 03:18:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18e07220cc9441d1ad5819ff7987a615 : SUCCESS in 6m 31s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/e1296dac1dfc4d5991a64a066c15efee : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a3d36f8e284f082615971d35740e021f467225e5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-09 06:00:16.000000000","message":"Patch Set 2:\n\nDid you do a release of diskimage-builder so that we can merge this?","accounts_in_message":[],"_revision_number":2},{"id":"a810f316ca364d861dd2d56f516d32f054bd1924","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-09 06:03:38.000000000","message":"Patch Set 2: Workflow-1\n\nneeds dib release","accounts_in_message":[],"_revision_number":2},{"id":"e95fc026a1fecae9271509832173e6bfaf06d2d6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-09 20:44:13.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"b9173090c2007d5da9eb94bf6ceeba56bb1400c0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-10 04:42:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c36532fda4e2c9acf0d7ea6df0fdfdecc960d5f","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-10 06:19:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4ee5f5d655e387cae30c83b961379aa4e3cb3dad","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-10 15:21:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef90a6005663728a77330221418010b5f9f7bf86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:21:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff0f337ce880288bff1e9b325fb8c46d9a815529","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:33:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"32218b1ba7a71915639b8bcafb5549e6722a0fd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:33:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c41e5b5acc624bcbb2a68ce02e3781b9 : SUCCESS in 6m 23s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/9a0cde32650e4ab49e7a7d5218f11f15 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5f807e6998e40b7eafb4fcf120c047d5239d052","revisions":{"ccd1c391b77e1802f694ad6948ed690b4f1f88d7":{"kind":"REWORK","_number":1,"created":"2019-10-04 06:11:49.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/24/686524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/686524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/686524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/686524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/686524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/686524/1"}}},"commit":{"parents":[{"commit":"72827398f5418b29f9b7f89793675545fddc6337","subject":"Add top-level nested virt labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72827398f5418b29f9b7f89793675545fddc6337"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-10-04 06:06:17.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-10-04 06:06:17.000000000","tz":600},"subject":"nodepool/elements : use abstracted commands","message":"nodepool/elements : use abstracted commands\n\nOn platforms such as CentOS 8 which are python3 first, \"virtualenv\"\nand \"pip\" may not exist (removed to avoid any confusion over them\nbeing v2 or v3 commands).\n\nThe referenced dib change exports new variables that should be correct\non all platforms for creating virtualenv\u0027s and pip installs.\n\n(note will require DIB release to be active on builders)\n\nDepends-On: https://review.opendev.org/684462\nChange-Id: I3414fb9e503f94ff744b560eff9ec0f4afdbb50e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccd1c391b77e1802f694ad6948ed690b4f1f88d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccd1c391b77e1802f694ad6948ed690b4f1f88d7"}]},"branch":"refs/heads/master"},"c5f807e6998e40b7eafb4fcf120c047d5239d052":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-09 03:07:27.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/24/686524/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/686524/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/686524/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/686524/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/686524/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/686524/2"}}},"commit":{"parents":[{"commit":"ed153e9e70d6160d8edfac41cee8c9508c31a85f","subject":"infra-package-needs: fix haveged install for all CentOS releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ed153e9e70d6160d8edfac41cee8c9508c31a85f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-10-04 06:06:17.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-10-09 03:07:08.000000000","tz":0},"subject":"nodepool/elements : use abstracted commands","message":"nodepool/elements : use abstracted commands\n\nOn platforms such as CentOS 8 which are python3 first, \"virtualenv\"\nand \"pip\" may not exist (removed to avoid any confusion over them\nbeing v2 or v3 commands).\n\nThe referenced dib change exports new variables that should be correct\non all platforms for creating virtualenv\u0027s and pip installs.\n\n(note will require DIB release to be active on builders)\n\nDepends-On: https://review.opendev.org/684462\nChange-Id: I3414fb9e503f94ff744b560eff9ec0f4afdbb50e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5f807e6998e40b7eafb4fcf120c047d5239d052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5f807e6998e40b7eafb4fcf120c047d5239d052"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
