)]}'
{"id":"openstack%2Fproject-config~687925","triplet_id":"openstack%2Fproject-config~master~Ic01915d85c0978bd962f9159f5a2ddfc7a53e787","project":"openstack/project-config","branch":"master","topic":"project-update","hashtags":[],"change_id":"Ic01915d85c0978bd962f9159f5a2ddfc7a53e787","subject":"Add OVN charms","status":"MERGED","created":"2019-10-10 15:00:18.000000000","updated":"2019-10-17 15:44:29.000000000","submitted":"2019-10-17 15:44:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687925-1571327069152-29eb8155","meta_rev_id":"3d261aeb2c28b4435a9c3af4439493d1a477d0d4","_number":687925,"virtual_id_number":687925,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 15:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-10-11 04:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-11 09:05:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-17 15:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-17 15:35:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 04:45:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-11 09:05:42.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2019-10-17 15:35:03.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-17 15:44:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1307717f8ad701f981a75ab7f362b62d7e59a169","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2019-10-10 15:00:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1213559c08ed29ae866ce75e2e3094fad18358c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-10 15:07:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4e0b0903f4d4ff6cceb31d75f0fe70cdbc917d0","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2019-10-10 15:17:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61c7842121e1b458e43007a7e76aa2a43f74b498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:30:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ddb7ced44c1e4e578979fc883998b02d : SUCCESS in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a5fae56e0d9c408e8248bb6c347f8c79 : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"948f9c52d823a943d79e9631e88d27cbd0604369","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2019-10-10 18:59:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d7eddaacf282dc6d0540ecc021035c7ac9e3f7b0","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2019-10-10 19:07:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2db39c22e5b330abdf9679a45b545c5fdc698d97","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2019-10-10 19:09:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0f01fe7310b1bd24c812422827317d2154a3d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:22:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a3eb836242546c083218a41b930e6b6 : SUCCESS in 7m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2344493fef2148109b41f52cf5648ec8 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":5},{"id":"f81e4ac864a2a9851e04139f6bfeeda3b758a0f4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-11 04:45:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"87f9383d260b98ca019278947c185b16065c4d58","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-10-11 09:05:42.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"871cb091fe164f3ce2d08b0ea4bb6de7534a4ce0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-17 15:35:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"290bb7b306b4e184ff5f78edb9fb6044f6898113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:35:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f513537faa3f05e8f3ec7a8cfd10b2ff8f28f10d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:44:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a06a11f1079a48b89965ee4c363b5601 : SUCCESS in 7m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2bc30f2425684803b4d6905cdb3e055c : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"44e8238a8abe7b0b5e81291827c396a0b228e95e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:44:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9663354b14744b3d72ba813c91e50779c4514055","revisions":{"985f198db0e55c118a704bd8d73bd8316327d1dd":{"kind":"REWORK","_number":1,"created":"2019-10-10 15:00:18.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/25/687925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/687925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/687925/1"}}},"commit":{"parents":[{"commit":"62b7a742f0cc7323b3c1775467ec727df2e03360","subject":"Merge \"gerritbot: notify stable branch changes to #openstack-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b7a742f0cc7323b3c1775467ec727df2e03360"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 14:56:00.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 14:59:24.000000000","tz":120},"subject":"Add OVN charms","message":"Add OVN charms\n\nNeeded-By: Ief1de4be550c852ba6daa9c5f25ff06a2dffb3db\nChange-Id: Ic01915d85c0978bd962f9159f5a2ddfc7a53e787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/985f198db0e55c118a704bd8d73bd8316327d1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/985f198db0e55c118a704bd8d73bd8316327d1dd"}]},"branch":"refs/heads/master"},"59bbfd4e38b35aee2f45100a14f9897161c65fb0":{"kind":"REWORK","_number":2,"created":"2019-10-10 15:17:08.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/25/687925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/687925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/687925/2"}}},"commit":{"parents":[{"commit":"62b7a742f0cc7323b3c1775467ec727df2e03360","subject":"Merge \"gerritbot: notify stable branch changes to #openstack-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b7a742f0cc7323b3c1775467ec727df2e03360"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 14:56:00.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 15:16:50.000000000","tz":120},"subject":"Add OVN charms","message":"Add OVN charms\n\nNeeded-By: Ief1de4be550c852ba6daa9c5f25ff06a2dffb3db\nChange-Id: Ic01915d85c0978bd962f9159f5a2ddfc7a53e787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59bbfd4e38b35aee2f45100a14f9897161c65fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59bbfd4e38b35aee2f45100a14f9897161c65fb0"}]},"branch":"refs/heads/master"},"c56c3f5666e86cd0d3a8be24c72b675a0fe966be":{"kind":"REWORK","_number":3,"created":"2019-10-10 18:59:22.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/25/687925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/687925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/687925/3"}}},"commit":{"parents":[{"commit":"62b7a742f0cc7323b3c1775467ec727df2e03360","subject":"Merge \"gerritbot: notify stable branch changes to #openstack-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b7a742f0cc7323b3c1775467ec727df2e03360"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 14:56:00.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 18:58:10.000000000","tz":120},"subject":"Add OVN charms","message":"Add OVN charms\n\nNeeded-By: Ief1de4be550c852ba6daa9c5f25ff06a2dffb3db\nChange-Id: Ic01915d85c0978bd962f9159f5a2ddfc7a53e787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c56c3f5666e86cd0d3a8be24c72b675a0fe966be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c56c3f5666e86cd0d3a8be24c72b675a0fe966be"}]},"branch":"refs/heads/master"},"cae2d57c069c9eb26c06cbf849ff4fd18dc11201":{"kind":"REWORK","_number":4,"created":"2019-10-10 19:07:07.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/25/687925/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/687925/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/687925/4"}}},"commit":{"parents":[{"commit":"62b7a742f0cc7323b3c1775467ec727df2e03360","subject":"Merge \"gerritbot: notify stable branch changes to #openstack-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b7a742f0cc7323b3c1775467ec727df2e03360"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 14:56:00.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 19:06:14.000000000","tz":120},"subject":"Add OVN charms","message":"Add OVN charms\n\nNeeded-By: Ief1de4be550c852ba6daa9c5f25ff06a2dffb3db\nChange-Id: Ic01915d85c0978bd962f9159f5a2ddfc7a53e787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cae2d57c069c9eb26c06cbf849ff4fd18dc11201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cae2d57c069c9eb26c06cbf849ff4fd18dc11201"}]},"branch":"refs/heads/master"},"9663354b14744b3d72ba813c91e50779c4514055":{"kind":"REWORK","_number":5,"created":"2019-10-10 19:09:38.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/25/687925/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/687925/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/687925/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/687925/5"}}},"commit":{"parents":[{"commit":"62b7a742f0cc7323b3c1775467ec727df2e03360","subject":"Merge \"gerritbot: notify stable branch changes to #openstack-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62b7a742f0cc7323b3c1775467ec727df2e03360"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 14:56:00.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@gmail.com","date":"2019-10-10 19:09:20.000000000","tz":120},"subject":"Add OVN charms","message":"Add OVN charms\n\nNeeded-By: Ief1de4be550c852ba6daa9c5f25ff06a2dffb3db\nChange-Id: Ic01915d85c0978bd962f9159f5a2ddfc7a53e787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9663354b14744b3d72ba813c91e50779c4514055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9663354b14744b3d72ba813c91e50779c4514055"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
