)]}'
{"id":"openstack%2Fproject-config~688443","triplet_id":"openstack%2Fproject-config~master~I5ac0f5ee7b8f18acfd9726e0fcec1c67bedde32d","project":"openstack/project-config","branch":"master","topic":"opensuse-423","hashtags":[],"change_id":"I5ac0f5ee7b8f18acfd9726e0fcec1c67bedde32d","subject":"Remove references to OpenSUSE 423","status":"MERGED","created":"2019-10-14 16:20:44.000000000","updated":"2019-10-16 04:56:39.000000000","submitted":"2019-10-16 04:56:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688443-1571201799084-33a7a8e1","meta_rev_id":"9439cd0739aa9b550a7d1e1603f2d0c34b2e2ac9","_number":688443,"virtual_id_number":688443,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 04:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-10-15 21:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-10-14 17:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-10-16 04:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 21:51:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-10-16 04:35:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-16 04:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2319df115535be46942abb2bf1283e551991dd4","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-14 16:20:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aabeee15eb2b6868ccfc482886bf8686adf239e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 16:34:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e2b83e5d538543da9b85d40a1b17a5f9 : SUCCESS in 6m 29s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/8226526208694710a69e81b830d8a774 : SUCCESS in 4m 11s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/a37c3e6220b748998765f1f66ccee584 : SUCCESS in 4m 08s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/f3ad0920d1b045619e026336b7f6294f : SUCCESS in 2m 53s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/1bdf861672344b2e8b95c6b83c3d7e50 : SUCCESS in 2m 24s\n- project-config-bindep-fallback-ubuntu-trusty https://zuul.opendev.org/t/openstack/build/bad3924781a040029ba3966b4b8b416c : SUCCESS in 3m 07s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/236ccafa7afe44de8fa180bcbfe140dc : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cb480ace5f5d2a5816f383930e69a783d7cebf5c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 17:35:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c9a72efb2511d7fa211f6771f9518efa3003a6e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-15 21:51:16.000000000","message":"Patch Set 1: Code-Review+2\n\nNot approving as it would be nice to have https://review.opendev.org/#/c/688442/ merge first.","accounts_in_message":[],"_revision_number":1},{"id":"367f34d84761c618a2a87a023c6ec99c8a45127f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-16 04:35:48.000000000","message":"Patch Set 1: Workflow+1\n\nnodeset got removed","accounts_in_message":[],"_revision_number":1},{"id":"2105908edfecd8937164d4e24649701ee06ff15b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:36:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7dc89be4b091cf3e1db90840478832bdffcc32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:56:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/71412b581c1345ca9eb0051b823e0648 : SUCCESS in 6m 03s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/afb31feddcdb46ddbe965bc9e2bb76ce : SUCCESS in 3m 11s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/71fe46afc1114540ab72c731ca2a4296 : SUCCESS in 3m 10s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/9a72e4494e6d4ba9a6ba7ddc53b0c942 : SUCCESS in 3m 24s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/fb694945648e480eafca0e6fca89b1c3 : SUCCESS in 2m 43s\n- project-config-bindep-fallback-ubuntu-trusty https://zuul.opendev.org/t/openstack/build/fafb33c41eff4eef8cbe6f173cd72dfa : SUCCESS in 4m 37s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b569aaa959f543379cebe3932369e76f : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"13fd0504f6e150a793a0ee16eec5784837b0a8ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:56:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1a1fe920370f553d7551d325e2ff21a437767ac","revisions":{"d1a1fe920370f553d7551d325e2ff21a437767ac":{"kind":"REWORK","_number":1,"created":"2019-10-14 16:20:44.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/43/688443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/688443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/688443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/688443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/688443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/688443/1"}}},"commit":{"parents":[{"commit":"d6a34dfd99cd8ee872be32498f8b001ba3b6d892","subject":"Merge \"Remove arm64ci-nrt1 (2/2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6a34dfd99cd8ee872be32498f8b001ba3b6d892"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-10-14 16:14:08.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2019-10-14 16:14:08.000000000","tz":-420},"subject":"Remove references to OpenSUSE 423","message":"Remove references to OpenSUSE 423\n\nThis nodeset is no longer in use, remove references to it.\n\nChange-Id: I5ac0f5ee7b8f18acfd9726e0fcec1c67bedde32d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1a1fe920370f553d7551d325e2ff21a437767ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1a1fe920370f553d7551d325e2ff21a437767ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
