)]}'
{"id":"openstack%2Fproject-config~688714","triplet_id":"openstack%2Fproject-config~master~I5687c25113c2f821c2d29df0c190b04b631f56c1","project":"openstack/project-config","branch":"master","topic":"barbican-ui","hashtags":[],"change_id":"I5687c25113c2f821c2d29df0c190b04b631f56c1","subject":"Add openstack/barbican-ui","status":"MERGED","created":"2019-10-15 13:53:40.000000000","updated":"2019-10-21 21:51:48.000000000","submitted":"2019-10-21 21:51:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688714-1571694708108-ed58b23b","meta_rev_id":"fff4dd5567e73a0bf718c5f570f8b6f4164495d7","_number":688714,"virtual_id_number":688714,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-10-15 15:18:59.000000000","_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 21:51:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-10-21 21:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-10-15 15:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2019-10-21 15:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-10-21 21:27:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 15:47:31.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-21 15:13:05.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-10-21 21:27:45.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-10-21 21:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eeaf21f866369610c45114c68167acfededbba26","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-10-15 13:53:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2626ba3f95c137942f79843b6c228420981068f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:06:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0c77fa1cba0a4a24acce41ab40c3d6f6 : SUCCESS in 6m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/baecb5888c6645dca4d62562953f9c23 : FAILURE in 3m 07s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c496b99123764f50ba68292fae4b5297 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c81edc3ab42b937ffde269cb184012505c6ff7f3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-15 14:38:48.000000000","message":"Patch Set 1: Workflow-1\n\nTesting branch origin/master\n  ERROR: Found .zuul.yaml on branch origin/master\n    Remove any zuul config files before import.","accounts_in_message":[],"_revision_number":1},{"id":"1f512e617f091635455a0c00687807a3f13b5f4b","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-10-15 14:50:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"87fc88c65cfb8c1310cbb3ebf33736513e52849c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 15:06:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/14107369363048c787030569776d46d2 : SUCCESS in 5m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8dcd7f83e75a4931b08bddf0ca233dd3 : SUCCESS in 4m 03s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e9d5cf6d73c34323b9be1ecbbc6c9519 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7c9fed0d6061ca031b10c46103e7ec8edf9518f1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-10-15 15:18:59.000000000","message":"Patch Set 1:\n\nThanks for the review, Andreas.  I\u0027ve removed the .zuul file from the repo and the gates are passing now.","accounts_in_message":[],"_revision_number":1},{"id":"c279319743631d7c68d77e625e0c3a7f1efdc521","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-15 15:47:31.000000000","message":"Patch Set 1: Code-Review+2 -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"b72ea41979c60574f219c54ad13f733c2875d69b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-21 15:13:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2d0d16c31e9519450ddf9d66da106675a281d27f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-21 21:27:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c4058194b77c1095dd83497b9af1a7d60105473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 21:28:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f61d7e1f1c41f7a78216b2d46604d308f4f0d027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 21:51:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dc341c68fcd8452d9ae28be84d89390e : SUCCESS in 6m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7004280b32154fd68fd176175f7125bd : SUCCESS in 3m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e0e3fba059a04285b35b8eefe63b7498 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d6a0d7ca0e456a539bc267e23ef2a222916c9787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 21:51:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35db482c5af4d5a67ca2e457a8b0e7955592ca69","revisions":{"35db482c5af4d5a67ca2e457a8b0e7955592ca69":{"kind":"REWORK","_number":1,"created":"2019-10-15 13:53:40.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/14/688714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/688714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/688714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/688714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/688714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/688714/1"}}},"commit":{"parents":[{"commit":"b9706c00097ecf3b97adad9e408293228c045799","subject":"Retire puppet-yum - remove gate jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b9706c00097ecf3b97adad9e408293228c045799"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-10-15 13:51:38.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-10-15 13:53:28.000000000","tz":-300},"subject":"Add openstack/barbican-ui","message":"Add openstack/barbican-ui\n\nNeeded-By: If6f64ae6e275e5e235ef0a41f5266fe209db3da4\n\nChange-Id: I5687c25113c2f821c2d29df0c190b04b631f56c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/35db482c5af4d5a67ca2e457a8b0e7955592ca69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/35db482c5af4d5a67ca2e457a8b0e7955592ca69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
