)]}'
{"id":"openstack%2Fproject-config~689220","triplet_id":"openstack%2Fproject-config~master~I2adbb575592eefe6958157d8fb007d0b50998e82","project":"openstack/project-config","branch":"master","topic":"remove-packethost","hashtags":[],"change_id":"I2adbb575592eefe6958157d8fb007d0b50998e82","subject":"Remove packethost provider","status":"MERGED","created":"2019-10-17 17:04:48.000000000","updated":"2019-10-17 17:56:06.000000000","submitted":"2019-10-17 17:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689220-1571334966024-14372de2","meta_rev_id":"5169390b3115708e9da7a41cd18f78cc58e6609e","_number":689220,"virtual_id_number":689220,"owner":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 17:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-10-17 17:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-10-17 17:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-17 17:46:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 17:18:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-10-17 17:34:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-17 17:46:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-10-17 17:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4ed01485fc2c9ec025a98054b5761d50b0c0267","author":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"date":"2019-10-17 17:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0aad6082a66fcbe7dc497c58ffa2146dacf362c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:14:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7ca93aa13e4a43cf955869c2bd032c49 : SUCCESS in 7m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0f812504552c46129940a6ed9ed7e59e : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"49118dc7d53865582a3ca61c26dc8574701467ef","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-17 17:18:23.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is easy enough to find and revert if the environment is ever resurrected.","accounts_in_message":[],"_revision_number":1},{"id":"b650488e927967319a71363c1112b75b93eb4356","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-17 17:27:10.000000000","message":"Patch Set 1: Code-Review-1\n\nLet\u0027s remove these as well:\ngrafana/create-nodepool.sh:create Packethost \u0027packethost-*\u0027 nodepool-packethost.yaml\ngrafana/nodepool-packethost.yaml:      query: stats.gauges.nodepool.provider.packethost-*","accounts_in_message":[],"_revision_number":1},{"id":"1a9de70e64db52c65c7cbb5f24da86d98318019c","author":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"date":"2019-10-17 17:30:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"203063c2f8fd97e54d9e2029171cd3404ac04281","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-17 17:34:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e9e368fbb0a3cb780425e7ae67c514f4bf59b910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:41:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f38d34dff92b434da748f4da3363f244 : SUCCESS in 7m 45s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/89a8927a51d34706a620036df4ef682d : SUCCESS in 2m 59s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e0cae69b72024e40805fe83d91f82925 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"7cde6a34f9ca0512203f76dcae85f50da092a567","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-17 17:46:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8a0659bfab71075d0e0ff9161f650f5211c3e826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:47:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a71969fe4d0efb9a641349c0a19f2b7b807705ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:56:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bd0a09ea61db426eb5945080a407c984 : SUCCESS in 5m 15s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/c5392b25bc444f1a92605ce2f1553b77 : SUCCESS in 3m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c12935f5f76c4814a4a41d504c00da89 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0d2c7b53e169ae1fad2b5b15048c80d6aff43cbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:56:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0278edda797d6033905136c17cec7627fcb2ee5","revisions":{"3a3004b6e4479a18baa3738971b8dac502a621e9":{"kind":"REWORK","_number":1,"created":"2019-10-17 17:04:48.000000000","uploader":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"ref":"refs/changes/20/689220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/689220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/689220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/689220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/689220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/689220/1"}}},"commit":{"parents":[{"commit":"e3259b7c64bb16d7f093ff0b50ad457c55ab63ec","subject":"Merge \"Add OVN charms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3259b7c64bb16d7f093ff0b50ad457c55ab63ec"}]}],"author":{"name":"David Shrewsbury","email":"dshrewsb@redhat.com","date":"2019-10-17 17:03:07.000000000","tz":-240},"committer":{"name":"David Shrewsbury","email":"dshrewsb@redhat.com","date":"2019-10-17 17:03:07.000000000","tz":-240},"subject":"Remove packethost provider","message":"Remove packethost provider\n\nThis provider has already been disabled in these config files by\nsetting max-servers to 0 and diskimages to []. There are errors\nin the nodepool logs about not being able to contact this provider\nnow since it appears disabled/gone, so time to remove the entries.\n\nChange-Id: I2adbb575592eefe6958157d8fb007d0b50998e82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3a3004b6e4479a18baa3738971b8dac502a621e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3a3004b6e4479a18baa3738971b8dac502a621e9"}]},"branch":"refs/heads/master"},"c0278edda797d6033905136c17cec7627fcb2ee5":{"kind":"REWORK","_number":2,"created":"2019-10-17 17:30:50.000000000","uploader":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"ref":"refs/changes/20/689220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/689220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/689220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/689220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/689220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/689220/2"}}},"commit":{"parents":[{"commit":"e3259b7c64bb16d7f093ff0b50ad457c55ab63ec","subject":"Merge \"Add OVN charms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3259b7c64bb16d7f093ff0b50ad457c55ab63ec"}]}],"author":{"name":"David Shrewsbury","email":"dshrewsb@redhat.com","date":"2019-10-17 17:03:07.000000000","tz":-240},"committer":{"name":"David Shrewsbury","email":"dshrewsb@redhat.com","date":"2019-10-17 17:30:45.000000000","tz":-240},"subject":"Remove packethost provider","message":"Remove packethost provider\n\nThis provider has already been disabled in these config files by\nsetting max-servers to 0 and diskimages to []. There are errors\nin the nodepool logs about not being able to contact this provider\nnow since it appears disabled/gone, so time to remove the entries.\n\nChange-Id: I2adbb575592eefe6958157d8fb007d0b50998e82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0278edda797d6033905136c17cec7627fcb2ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0278edda797d6033905136c17cec7627fcb2ee5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
