)]}'
{"id":"openstack%2Fproject-config~689579","triplet_id":"openstack%2Fproject-config~master~Ib82552520b99ffcf213962e917315b0b82c3f1b0","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ib82552520b99ffcf213962e917315b0b82c3f1b0","subject":"Update nodepool element for disk cache update","status":"MERGED","created":"2019-10-18 20:35:18.000000000","updated":"2019-10-18 20:58:15.000000000","submitted":"2019-10-18 20:58:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689579-1571432295009-33a7a8e1","meta_rev_id":"9aec35f94f74c62924b369fa771208f5ca7c2d28","_number":689579,"virtual_id_number":689579,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 20:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-10-18 20:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 20:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 20:36:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 20:36:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-10-18 20:42:45.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-10-18 20:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1e15592ff9f14258b7559c61118f4b2a5f849cc","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-18 20:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d6324b4b5bc317f85ebf24f7351ead8769361d1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-18 20:35:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b8e57a031a4b0577707ba943000c520d3aff27d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-18 20:36:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"89ae28f837356dfebda369ea56f97fbf3715d4ef","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-18 20:37:17.000000000","message":"Patch Set 2:\n\nSingle-core approving emergency patch to get all images building again.","accounts_in_message":[],"_revision_number":2},{"id":"b9cc9804108cf20a61b397f655bba1dda8fcece6","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-18 20:42:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0a14457011c1eb30cf3a81637824a0a2ab174a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 20:46:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6941dff3b84e4af6af69f177c4e557fc : SUCCESS in 5m 54s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/bedbd3b961c94f8f8247fe84a33c7335 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"00c4b06355c121503f3562c99b438778b21d2b75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 20:46:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6602458822d536efde8419d06f0f67f1f99b5372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 20:58:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bdbdb21fc8cd4f88ba415f126a78062f : SUCCESS in 6m 27s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/51f27cf3e83847109352c26762b551f3 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9aec35f94f74c62924b369fa771208f5ca7c2d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 20:58:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"66b12be818bf678ba72c8f9f374d301e40e7bfd2","revisions":{"887507cc466461732c3f5dcfc6544483e86ab561":{"kind":"REWORK","_number":1,"created":"2019-10-18 20:35:18.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/79/689579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/689579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/689579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/689579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/689579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/689579/1"}}},"commit":{"parents":[{"commit":"fe98ca095e33b1902b515fa49c05cd5cdc304117","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe98ca095e33b1902b515fa49c05cd5cdc304117"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-10-18 20:34:55.000000000","tz":540},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-10-18 20:34:55.000000000","tz":540},"subject":"Update nodepool element for disk cache update","message":"Update nodepool element for disk cache update\n\nThings are in /opt/git/opendev.org/ now\n\nChange-Id: Ib82552520b99ffcf213962e917315b0b82c3f1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/887507cc466461732c3f5dcfc6544483e86ab561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/887507cc466461732c3f5dcfc6544483e86ab561"}]},"branch":"refs/heads/master"},"66b12be818bf678ba72c8f9f374d301e40e7bfd2":{"kind":"REWORK","_number":2,"created":"2019-10-18 20:35:52.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/79/689579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/689579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/689579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/689579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/689579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/689579/2"}}},"commit":{"parents":[{"commit":"fe98ca095e33b1902b515fa49c05cd5cdc304117","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe98ca095e33b1902b515fa49c05cd5cdc304117"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-10-18 20:34:55.000000000","tz":540},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-10-18 20:35:46.000000000","tz":540},"subject":"Update nodepool element for disk cache update","message":"Update nodepool element for disk cache update\n\nThings are in /opt/git/opendev.org/ now\n\nChange-Id: Ib82552520b99ffcf213962e917315b0b82c3f1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66b12be818bf678ba72c8f9f374d301e40e7bfd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66b12be818bf678ba72c8f9f374d301e40e7bfd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
