)]}'
{"id":"openstack%2Fproject-config~689895","triplet_id":"openstack%2Fproject-config~master~Ia9e8057d1f53901113702f49fb24e1733f368362","project":"openstack/project-config","branch":"master","topic":"pause-fedora","hashtags":[],"change_id":"Ia9e8057d1f53901113702f49fb24e1733f368362","subject":"Pause Fedora 30 builds","status":"MERGED","created":"2019-10-21 23:18:18.000000000","updated":"2019-10-22 00:49:13.000000000","submitted":"2019-10-22 00:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689895-1571705353348-bfc748b9","meta_rev_id":"2bd51b1a2a700ead44553aaa0379dfbb756edddb","_number":689895,"virtual_id_number":689895,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 00:49:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-10-21 23:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-22 00:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 23:22:41.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-10-22 00:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a48e1bdeeba0c93ffce56f7323bfd5bbc5993663","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-21 23:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"018da949f094f54d0ef945d661a006ef7091836c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-21 23:22:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f7a1c0fa867b9772894ee49618d55e0c14d77a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 23:26:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e69cd533efa848f3b9b92baa9d2f6296 : SUCCESS in 6m 23s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/5248b799094443469b74f92861236ca6 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c93d07650a32714b34ca88a008847dfc35c2106a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-10-22 00:42:00.000000000","message":"Patch Set 1: Workflow+1\n\napproving with one review and will monitor; but this has been failing for days so no point in keeping it looping","accounts_in_message":[],"_revision_number":1},{"id":"df61062eb4de5d01384014813592bd3373e3fe2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 00:42:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"054ff439bd3aedd5e94ff1c8e82f986c0d3b1344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 00:49:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6531eb7b6359447496310f6dea5c10f6 : SUCCESS in 5m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ff7cf76915204a42bd058ad0deb4b392 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"12837e91de71440c4c854f70ddf923ae92253ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 00:49:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1e8e7bfb4cb96f68775127dad8610bde1d333ee","revisions":{"a1e8e7bfb4cb96f68775127dad8610bde1d333ee":{"kind":"REWORK","_number":1,"created":"2019-10-21 23:18:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/689895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/689895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/689895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/689895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/689895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/689895/1"}}},"commit":{"parents":[{"commit":"62af5406b100650c90fb0512a4410aced531696a","subject":"Merge \"Add openstack/barbican-ui\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62af5406b100650c90fb0512a4410aced531696a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-10-21 23:17:16.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-10-21 23:17:16.000000000","tz":660},"subject":"Pause Fedora 30 builds","message":"Pause Fedora 30 builds\n\nFedora 30 paused until dnf segfaults resolved; see\n\n https://bugzilla.redhat.com/show_bug.cgi?id\u003d1758737\n https://bugzilla.redhat.com/show_bug.cgi?id\u003d1761282\n\nChange-Id: Ia9e8057d1f53901113702f49fb24e1733f368362\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1e8e7bfb4cb96f68775127dad8610bde1d333ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1e8e7bfb4cb96f68775127dad8610bde1d333ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
