)]}'
{"id":"openstack%2Fproject-config~691038","triplet_id":"openstack%2Fproject-config~master~I7d0d37dcc4c9cded6b2d48650097032b819a0634","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I7d0d37dcc4c9cded6b2d48650097032b819a0634","subject":"Update Project Configs for x/ranger","status":"MERGED","created":"2019-10-24 18:15:40.000000000","updated":"2019-10-25 21:16:55.000000000","submitted":"2019-10-25 21:16:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691038-1572038215978-b61b701d","meta_rev_id":"08cdc7faacf8fd043a8b6e109a54896690b8101b","_number":691038,"virtual_id_number":691038,"owner":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},{"value":0,"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 21:16:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},"all":[{"value":1,"date":"2019-10-25 19:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},{"value":2,"date":"2019-10-25 21:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},{"value":1,"date":"2019-10-24 20:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 18:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},{"value":1,"date":"2019-10-25 21:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},{"value":0,"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},{"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 19:15:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-24 20:20:56.000000000","updated_by":{"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},"reviewer":{"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},"state":"REVIEWER"},{"updated":"2019-10-25 18:54:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-10-25 19:25:32.000000000","updated_by":{"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},"reviewer":{"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},"state":"REVIEWER"},{"updated":"2019-10-25 21:02:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-10-25 21:16:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d61ee14734fac5ee5a58932626eba24056c019fd","author":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"date":"2019-10-24 18:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1eee66b5467fbdb574b355d86f8e253aff7b3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:29:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/92f0c7ad71f0435abd24dbe3f710beb9 : SUCCESS in 6m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/68ac80a6e8334749b4ada38955ea3099 : FAILURE in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"af271dacde7e16179c61933642c5b1d49898257a","author":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"date":"2019-10-24 18:59:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ad51a2edf39b91354d7b25d28788ebdbcf5aa07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:12:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/941cdd32561d47c3aaf26f2ebb7b2a97 : SUCCESS in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0966da27552a4833a9dd3f8f10481347 : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ebb7d32507f8041fb0eeca01ba56878721e10f48","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-24 19:15:56.000000000","message":"Patch Set 2:\n\nwhat problem are you trying to fix really?\n\nNote that the file you modify is sorted alphabetical","accounts_in_message":[],"_revision_number":2},{"id":"2946ae3edd173a84c60e03b325978c9a9791eddb","author":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"date":"2019-10-24 19:57:39.000000000","message":"Patch Set 2:\n\nAndreas, thank you for  your input.  I\u0027ll alphabetize the order and re-commit.","accounts_in_message":[],"_revision_number":2},{"id":"f563d697855edafd58663faade4a68f1b1d01976","author":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"date":"2019-10-24 19:58:39.000000000","message":"Patch Set 2:\n\nWe are trying to fix the following issue: \n[remote rejected] HEAD -\u003e refs/publish/master (you are not allowed to upload merges)","accounts_in_message":[],"_revision_number":2},{"id":"8753fb4e2ba3d573bf97a0e00d9d4b8b83507249","author":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"date":"2019-10-24 19:59:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"09ed67a25d289af04caf6a9f7f3c3c447d035b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:06:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7af4f37360c949e0977c75f64f68df4d : SUCCESS in 5m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/752bcd016149455d8935d068ddaefd5a : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a8ba545987cd8579756a502b9855434524b2ddca","author":{"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},"date":"2019-10-24 20:20:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cef5b28b56f9165a3b1373f7cd2fe6c8f46f50cd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-25 16:02:02.000000000","message":"Patch Set 3:\n\nAre you trying to propose/review merge commits between branches (e.g., between a feature branch and master)? The default ACLs don\u0027t include pushMerge mainly because most of the time when people get that error it\u0027s that they\u0027re accidentally trying to submit changes from a dirty local checkout and would end up polluting the resulting review (and possibly the target branch if that change were then to get merged).","accounts_in_message":[],"_revision_number":3},{"id":"cd42a21674bed9553f6a4dbc77579110a916113e","author":{"_account_id":24999,"name":"Hari","email":"hosingh000@gmail.com","username":"hosingh000"},"date":"2019-10-25 16:15:20.000000000","message":"Patch Set 3:\n\n\u003e Are you trying to propose/review merge commits between branches\n \u003e (e.g., between a feature branch and master)? The default ACLs don\u0027t\n \u003e include pushMerge mainly because most of the time when people get\n \u003e that error it\u0027s that they\u0027re accidentally trying to submit changes\n \u003e from a dirty local checkout and would end up polluting the\n \u003e resulting review (and possibly the target branch if that change\n \u003e were then to get merged).\n\nYes we have some branche which changes needed to be merged in master.","accounts_in_message":[],"_revision_number":3},{"id":"9c488c52d95cdfac90d3385f0d67ea632c09807e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-25 18:54:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5e4a22551b0d332e9a3a5645020ce84dfcbef0e5","author":{"_account_id":29585,"name":"Jeremy Houser","email":"jeremyhouser@protonmail.com","username":"jh629g"},"date":"2019-10-25 19:25:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7edf2f74490d004c480b14f15380019a2def3b3c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-25 21:02:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c023555abecb12c3723a298ff022fdd3e7cb73a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 21:02:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f1a9a9ba86a4b15c0450525a39a61cd39e047ed9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 21:16:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"08cdc7faacf8fd043a8b6e109a54896690b8101b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 21:16:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0f2c4f4686bc49ceaad48fc5364b4c17 : SUCCESS in 8m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/720eb4ceae1b487bb40c3f4f952fe5b3 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5a63740a94be049277deb500718370d3a13cd970","revisions":{"aad7bced4fc45f0a335615e39ae670bd4dd7c01e":{"kind":"REWORK","_number":1,"created":"2019-10-24 18:15:40.000000000","uploader":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"ref":"refs/changes/38/691038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/691038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/691038/1"}}},"commit":{"parents":[{"commit":"3ab747071b4e1faca557bc59861d2ed41bfb9fc1","subject":"Merge \"Enable inap-mtl01\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ab747071b4e1faca557bc59861d2ed41bfb9fc1"}]}],"author":{"name":"stewie925","email":"st3wty@att.com","date":"2019-10-24 18:09:44.000000000","tz":-420},"committer":{"name":"stewie925","email":"st3wty@att.com","date":"2019-10-24 18:09:44.000000000","tz":-420},"subject":"Update Project Configs for x/ranger","message":"Update Project Configs for x/ranger\n\nModify x/ranger config to fix the upload merge issue\n\nChange-Id: I7d0d37dcc4c9cded6b2d48650097032b819a0634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aad7bced4fc45f0a335615e39ae670bd4dd7c01e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aad7bced4fc45f0a335615e39ae670bd4dd7c01e"}]},"branch":"refs/heads/master"},"3575ab1c9fdee36f3fb38a6e45a4ffa1b11aadba":{"kind":"REWORK","_number":2,"created":"2019-10-24 18:59:07.000000000","uploader":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"ref":"refs/changes/38/691038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/691038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/691038/2"}}},"commit":{"parents":[{"commit":"3ab747071b4e1faca557bc59861d2ed41bfb9fc1","subject":"Merge \"Enable inap-mtl01\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ab747071b4e1faca557bc59861d2ed41bfb9fc1"}]}],"author":{"name":"stewie925","email":"st3wty@att.com","date":"2019-10-24 18:09:44.000000000","tz":-420},"committer":{"name":"STEW TY","email":"st3wty@att.com","date":"2019-10-24 18:58:26.000000000","tz":0},"subject":"Update Project Configs for x/ranger","message":"Update Project Configs for x/ranger\n\nModify x/ranger config to fix the upload merge issue\n\nChange-Id: I7d0d37dcc4c9cded6b2d48650097032b819a0634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3575ab1c9fdee36f3fb38a6e45a4ffa1b11aadba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3575ab1c9fdee36f3fb38a6e45a4ffa1b11aadba"}]},"branch":"refs/heads/master"},"5a63740a94be049277deb500718370d3a13cd970":{"kind":"REWORK","_number":3,"created":"2019-10-24 19:59:39.000000000","uploader":{"_account_id":21111,"name":"STEW TY","email":"punsoca@gmail.com","username":"stewie925"},"ref":"refs/changes/38/691038/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/691038/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/691038/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/691038/3"}}},"commit":{"parents":[{"commit":"3ab747071b4e1faca557bc59861d2ed41bfb9fc1","subject":"Merge \"Enable inap-mtl01\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ab747071b4e1faca557bc59861d2ed41bfb9fc1"}]}],"author":{"name":"stewie925","email":"st3wty@att.com","date":"2019-10-24 18:09:44.000000000","tz":-420},"committer":{"name":"STEW TY","email":"st3wty@att.com","date":"2019-10-24 19:59:34.000000000","tz":0},"subject":"Update Project Configs for x/ranger","message":"Update Project Configs for x/ranger\n\nModify x/ranger config to fix the upload merge issue\n\nChange-Id: I7d0d37dcc4c9cded6b2d48650097032b819a0634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a63740a94be049277deb500718370d3a13cd970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a63740a94be049277deb500718370d3a13cd970"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
