)]}'
{"id":"openstack%2Fproject-config~691140","triplet_id":"openstack%2Fproject-config~master~Id201b8c7f624287021492daa80b4bd2f57efc855","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Id201b8c7f624287021492daa80b4bd2f57efc855","subject":"Grant permissions for murano-dashboard to horizon-core","status":"MERGED","created":"2019-10-25 01:51:59.000000000","updated":"2019-11-15 05:54:59.000000000","submitted":"2019-11-15 05:54:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691140-1573797299516-a3a83055","meta_rev_id":"eb40f39ca50ad9c5647d23277c9e9313c32f4e3b","_number":691140,"virtual_id_number":691140,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 05:54:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-11-13 07:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2019-11-13 10:33:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-11-13 08:29:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-15 05:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-15 05:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 07:22:31.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-05 09:32:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-11-13 07:34:24.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-13 08:29:11.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-13 10:33:53.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-11-15 05:44:35.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-11-15 05:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fabe57eca40f92d3b21854030a562100223fbf9","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-10-25 01:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcdebfcc5bee79bddc73e73214b142053cad8776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 01:58:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/856139e53ff444a695ad803d16a3f148 : SUCCESS in 5m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/59b93b902c2241f59521677827698c63 : FAILURE in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"74ce5a4beccd44a32cd4e5c7b84952db2adaa448","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-10-25 02:04:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9f564f6e6b2b74ef8133287a89e8a898d86c8255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 02:17:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9705b585aabe479b811d3c4361a50a99 : SUCCESS in 6m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0d72e83021a1412eac8cce1c8ab146c9 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"82fb81baf1c1669319ee83deb539d02082922f88","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-25 04:53:41.000000000","message":"Patch Set 2: Code-Review-1\n\ndon\u0027t you need refs/heads/stable/ ?","accounts_in_message":[],"_revision_number":2},{"id":"d6691b0f4d9682b7bdcf81fc28f3ef4e73f11178","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-10-25 07:26:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2ea72765d4a75dd0bc2809cbc60aea102764c3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 07:51:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5cd0dc63ac5b431e9afc664196ec0d74 : SUCCESS in 8m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/22132004829e4715b4b1c0be8e934f6f : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"aba9fa5988a42614e1ae73bf449443b87df36790","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-25 14:46:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"337a39ff1c5c10981c5f1c155036901aff5e6c5d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-10-26 07:44:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5e9accc76dccb98be96eb5b90c230edace345ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 07:51:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5395f47a18ff494cab34f8ad741115ca : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4a49aa991ad94df0ba8f3028acf0b9b5 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5305b1eb76cd2ab2200cf034a12e86aa9fb207c5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-26 18:53:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"263bdb61b9305d70192337f68b0f728c6e5c2e72","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-28 07:22:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1bdc8748e9ff24113cdb6c0dc60542e7711cdedf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-05 09:32:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7fc9ef864f683367be5f67bbe878d395893fa37b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-05 10:13:59.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"e077eb5d394c1b0c206e9e5c138c42283fd79b9c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-05 10:15:09.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b273787b1cc017a626ef003f9d2d7991aecbae44","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-11-13 00:51:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2f7c1f9e111ee4983cedd99c386bef883dffa3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 00:58:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/69de8aa84c9945f6bf903a5683b73c35 : SUCCESS in 6m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a1eb1dbc741c4a16a1685064dd18cb6e : FAILURE in 3m 11s","accounts_in_message":[],"_revision_number":5},{"id":"2f962ca8190d07fa69fb14d9680bb9bf9924cb9c","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-11-13 01:06:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"09c7e6af256ee9696ce43f681283b072d8e0208c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 01:11:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/054c6506c15e42eca08c560ae6330677 : SUCCESS in 5m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9d6ea1193ad143d79c1f77eb9e87fb09 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":6},{"id":"673a6e30f658b593d83d5c96fbbd427acbb9a3da","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-13 07:34:24.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"fd6918fe4370ccd8fa5af617d2a043fdbc88ce72","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 08:29:11.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"27372fc5f83cd551cd65b9e75b75d09acac36310","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-11-13 10:33:53.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e0509ddc56d79f377bdda66ea4e2b21b19eccdc8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-11-15 05:44:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d85d96beccfea9667c2cc232fc6462e1c8f43aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:44:48.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b838f7f40fc86643be4dff2e3b7d6ec84ca5acf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:54:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"eb40f39ca50ad9c5647d23277c9e9313c32f4e3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:54:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/05ef95d2ac7a41b4a69e65172e4116f4 : SUCCESS in 5m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6451338f85a44ad4a86f7cfd4d7cd913 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eb2bdfaf945ca835ff08caa8faefe244aac5f934","revisions":{"d969b6792e20409f81aea0dcfea16669926eeff5":{"kind":"REWORK","_number":1,"created":"2019-10-25 01:51:59.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/40/691140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/691140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/691140/1"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"subject":"Grant permissions for murano-dashboard to horizon-core","message":"Grant permissions for murano-dashboard to horizon-core\n\nChange-Id: Id201b8c7f624287021492daa80b4bd2f57efc855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d969b6792e20409f81aea0dcfea16669926eeff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d969b6792e20409f81aea0dcfea16669926eeff5"}]},"branch":"refs/heads/master"},"9461853bfb7e3ac99ec176c0a07b29f278e10d30":{"kind":"REWORK","_number":2,"created":"2019-10-25 02:04:23.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/40/691140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/691140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/691140/2"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 02:04:19.000000000","tz":0},"subject":"Grant permissions for murano-dashboard to horizon-core","message":"Grant permissions for murano-dashboard to horizon-core\n\nChange-Id: Id201b8c7f624287021492daa80b4bd2f57efc855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9461853bfb7e3ac99ec176c0a07b29f278e10d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9461853bfb7e3ac99ec176c0a07b29f278e10d30"}]},"branch":"refs/heads/master"},"b75d19151b9519ea1ee533b0f7590d48c608bf7e":{"kind":"REWORK","_number":3,"created":"2019-10-25 07:26:33.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/40/691140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/691140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/691140/3"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 07:26:29.000000000","tz":0},"subject":"Grant permissions for murano-dashboard to horizon-core","message":"Grant permissions for murano-dashboard to horizon-core\n\nChange-Id: Id201b8c7f624287021492daa80b4bd2f57efc855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b75d19151b9519ea1ee533b0f7590d48c608bf7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b75d19151b9519ea1ee533b0f7590d48c608bf7e"}]},"branch":"refs/heads/master"},"60fdf6364ad7e512d61bf3d077bbc00811c88ac6":{"kind":"REWORK","_number":4,"created":"2019-10-26 07:44:15.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/40/691140/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/691140/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/691140/4"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-26 07:44:12.000000000","tz":0},"subject":"Grant permissions for murano-dashboard to horizon-core","message":"Grant permissions for murano-dashboard to horizon-core\n\nChange-Id: Id201b8c7f624287021492daa80b4bd2f57efc855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60fdf6364ad7e512d61bf3d077bbc00811c88ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60fdf6364ad7e512d61bf3d077bbc00811c88ac6"}]},"branch":"refs/heads/master"},"aec963c0d2a160a56a9fc95a0113d0f96c98cde8":{"kind":"REWORK","_number":5,"created":"2019-11-13 00:51:15.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/40/691140/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/691140/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/691140/5"}}},"commit":{"parents":[{"commit":"eb699621606db5f4a0f8ea060781a289af1cf4b5","subject":"Merge \"Remove python27 jobs from neutron dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb699621606db5f4a0f8ea060781a289af1cf4b5"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-11-13 00:50:53.000000000","tz":-480},"subject":"Grant permissions for murano-dashboard to horizon-core","message":"Grant permissions for murano-dashboard to horizon-core\n\nChange-Id: Id201b8c7f624287021492daa80b4bd2f57efc855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aec963c0d2a160a56a9fc95a0113d0f96c98cde8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aec963c0d2a160a56a9fc95a0113d0f96c98cde8"}]},"branch":"refs/heads/master"},"eb2bdfaf945ca835ff08caa8faefe244aac5f934":{"kind":"REWORK","_number":6,"created":"2019-11-13 01:06:05.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/40/691140/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/691140/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/691140/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/691140/6"}}},"commit":{"parents":[{"commit":"eb699621606db5f4a0f8ea060781a289af1cf4b5","subject":"Merge \"Remove python27 jobs from neutron dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb699621606db5f4a0f8ea060781a289af1cf4b5"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-10-25 01:51:35.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-11-13 01:05:44.000000000","tz":-480},"subject":"Grant permissions for murano-dashboard to horizon-core","message":"Grant permissions for murano-dashboard to horizon-core\n\nChange-Id: Id201b8c7f624287021492daa80b4bd2f57efc855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb2bdfaf945ca835ff08caa8faefe244aac5f934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb2bdfaf945ca835ff08caa8faefe244aac5f934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
