)]}'
{"id":"openstack%2Fproject-config~691275","triplet_id":"openstack%2Fproject-config~master~I5e2c4b6a056b50077ab343535a4bbe3bc49a0acf","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I5e2c4b6a056b50077ab343535a4bbe3bc49a0acf","subject":"Bump min-ready for ubuntu-bionic","status":"MERGED","created":"2019-10-25 09:56:12.000000000","updated":"2019-10-25 22:38:53.000000000","submitted":"2019-10-25 22:38:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691275-1572043133012-89326a6a","meta_rev_id":"2b5e30e890a22c1f092f45d0c0d2341bf6176b0d","_number":691275,"virtual_id_number":691275,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 22:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-10-25 22:23:51.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 22:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-10-25 22:23:51.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 09:56:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-25 22:09:11.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-10-25 22:23:51.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-10-25 22:24:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-10-25 22:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b4a2bbeffd413dbb0df802c76cb1465c12448ba","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-10-25 09:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8320b9bc53daf7249713071a51d5093ee6a026f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:04:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/90c52ef74846459daed3fadbdb20865e : SUCCESS in 6m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ee0c6957f7794203bcc112eb5c5601a6 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0aba434f9063433ec7ecb24a77d2791f07510fe0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-25 22:09:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nI\u0027ve suggested an alternative inline that I would prefer but this should be fine too.","accounts_in_message":[],"_revision_number":1},{"id":"80df9ca6db289a88aaba26d7fe42b0e1adb01164","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-25 22:14:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59315ea9609b727eb39b58eaebb6c8322e2daf5d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-25 22:19:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d61cc18bb5bfa477e4aff8022347f80608742f47","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-25 22:21:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b60ef3981f8af4d390e9556146b0a8e218ea2c2b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-25 22:23:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06bf45cd86d47e6e5c1a973279b8101cc541d62f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-25 22:24:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a2c5970262deaed7247192dd648574c269a0cd14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 22:29:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/19410428359e47c4a81e42b20e186a93 : SUCCESS in 5m 39s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/da07aa8233854c7786987ae54e43a38d : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"59c35431853266d2ba30c0d633b692aed8062e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 22:29:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c939cda0f00a49db2d786f5b28633b75c3d2601f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 22:38:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ceb1bd16a29d4f10ad8b5bd2f0d0cc81 : SUCCESS in 7m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9ec975784fbb4061b204eee6e4e5756f : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"2b5e30e890a22c1f092f45d0c0d2341bf6176b0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 22:38:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7cda9f926681981c43195161607a488737765f3","revisions":{"1d5ec5cf135cf776505cd9b59d00ad699062cc2a":{"kind":"REWORK","_number":1,"created":"2019-10-25 09:56:12.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/75/691275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/691275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/691275/1"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-25 09:54:40.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-25 09:54:40.000000000","tz":60},"subject":"Bump min-ready for ubuntu-bionic","message":"Bump min-ready for ubuntu-bionic\n\nRaise min-ready from 1 to 3 for ubuntu-bionic after observing that\nlinters jobs used to start considerably later than more heavy ones.\n\nWhen looking at them, I discovered that they were using ubuntu-bionic\nand our current configuration explains why this was happening.\n\nChange-Id: I5e2c4b6a056b50077ab343535a4bbe3bc49a0acf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d5ec5cf135cf776505cd9b59d00ad699062cc2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d5ec5cf135cf776505cd9b59d00ad699062cc2a"}]},"branch":"refs/heads/master"},"60a2ef7ffe0c663b44f694aadcf30cf99182b33a":{"kind":"REWORK","_number":2,"created":"2019-10-25 22:19:48.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/75/691275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/691275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/691275/2"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-25 09:54:40.000000000","tz":60},"committer":{"name":"Clark Boylan","email":"cboylan@sapwetik.org","date":"2019-10-25 22:19:42.000000000","tz":0},"subject":"Bump min-ready for ubuntu-bionic","message":"Bump min-ready for ubuntu-bionic\n\nRaise min-ready from 1 to 3 for ubuntu-bionic after observing that\nlinters jobs used to start considerably later than more heavy ones.\n\nWhen looking at them, I discovered that they were using ubuntu-bionic\nand our current configuration explains why this was happening.\n\nChange-Id: I5e2c4b6a056b50077ab343535a4bbe3bc49a0acf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60a2ef7ffe0c663b44f694aadcf30cf99182b33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60a2ef7ffe0c663b44f694aadcf30cf99182b33a"}]},"branch":"refs/heads/master"},"a7cda9f926681981c43195161607a488737765f3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-25 22:21:52.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/75/691275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/691275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/691275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/691275/3"}}},"commit":{"parents":[{"commit":"ca400e34e9934674a403409bdff2c838401b9c63","subject":"arm64 images: add journal-to-console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca400e34e9934674a403409bdff2c838401b9c63"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-25 09:54:40.000000000","tz":60},"committer":{"name":"Clark Boylan","email":"cboylan@sapwetik.org","date":"2019-10-25 22:21:49.000000000","tz":0},"subject":"Bump min-ready for ubuntu-bionic","message":"Bump min-ready for ubuntu-bionic\n\nFlip the min ready values for Xenial and Bionic. This should\u0027ve been\ndone when we did the initial transition from Xenial to Bionic but was\nmissed. We want to have ready nodes for the common up to date ubuntu\nplatform (Bionic) and don\u0027t as many on the previous release (Xenial).\n\nThank you to zbr for catching this after noticing that linter jobs took\na while to start.\n\nChange-Id: I5e2c4b6a056b50077ab343535a4bbe3bc49a0acf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7cda9f926681981c43195161607a488737765f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7cda9f926681981c43195161607a488737765f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
