)]}'
{"id":"openstack%2Fproject-config~693643","triplet_id":"openstack%2Fproject-config~master~Ibcaddc259b3a92db4e07025c2c8331437e0cec61","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ibcaddc259b3a92db4e07025c2c8331437e0cec61","subject":"Use container instead of deboostrap for ubuntu-bionic","status":"ABANDONED","created":"2019-11-11 02:17:03.000000000","updated":"2021-09-03 12:14:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84cf322d445bd7df60d0c2f44cb6e934fd22964b","_number":693643,"virtual_id_number":693643,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-11 02:26:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 02:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96ce8b6cde8c8a1f9b56adaffbf529585ff71195","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-11-11 02:17:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7edbb95721a737f06c66ed9d111aeb206aab6d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 02:26:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6a281c9deabd4d7284dc9320673607c4 : SUCCESS in 5m 48s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bf85546b4a1f4deabf1bc77e038c47a4 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"84cf322d445bd7df60d0c2f44cb6e934fd22964b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2021-09-03 12:14:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78bb12994a5368d41630fd1d6893b2b0f73ce915","revisions":{"78bb12994a5368d41630fd1d6893b2b0f73ce915":{"kind":"REWORK","_number":1,"created":"2019-11-11 02:17:03.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/43/693643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/693643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/693643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/693643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/693643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/693643/1"}}},"commit":{"parents":[{"commit":"3348ed932ed3f5ba4fefc3145c913141ca6c9b09","subject":"Merge \"Update neutron-tempest-plugin grafana dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3348ed932ed3f5ba4fefc3145c913141ca6c9b09"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-11-11 02:16:05.000000000","tz":480},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-11-11 02:16:05.000000000","tz":480},"subject":"Use container instead of deboostrap for ubuntu-bionic","message":"Use container instead of deboostrap for ubuntu-bionic\n\nMigrate from using debootstrap to using the container image\npublished to dockerhub.\n\nDepends-On: https://review.opendev.org/693642\nChange-Id: Ibcaddc259b3a92db4e07025c2c8331437e0cec61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/78bb12994a5368d41630fd1d6893b2b0f73ce915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/78bb12994a5368d41630fd1d6893b2b0f73ce915"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
